)]}'
{"id":"openstack%2Fswift~810826","triplet_id":"openstack%2Fswift~stable%2Fxena~Ib67d08cbde3f952ae40a549d8458bf1d2dd99c7f","project":"openstack/swift","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"Ib67d08cbde3f952ae40a549d8458bf1d2dd99c7f","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","status":"MERGED","created":"2021-09-24 07:23:50.000000000","updated":"2021-09-30 03:33:04.000000000","submitted":"2021-09-30 03:31:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810826-create-xena","meta_rev_id":"4bb8aa1b708623ef9afe6392db50d015c2b6c1ac","_number":810826,"virtual_id_number":810826,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 03:31:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-30 02:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-30 02:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-24 08:22:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 02:13:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"ceb6d58c2c68f4309e9b4eb67b4165ea486220ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-24 07:23:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96c435c87a2d6b3df7d7513cf494824e696c8c2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 08:22:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0e207c379b41d08b6bdcf753192dff : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/934ed1ca476a4075b277c18602bdd833 : SUCCESS in 7m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/068de5eb206842bd8f70a83f5c891ce2 : FAILURE in 7m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2e189b4917b7419ea1813500c28d065f : FAILURE in 8m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9f6e4af7ff7c4990952f575691d65622 : SUCCESS in 20m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28573a6e1e1b4e78977dacab71e57d0f : SUCCESS in 25m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c3d5a3c204f4f09bcab93d52813f1dc : SUCCESS in 21m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f67a493cf6b544c99ecca6b40d3abe0b : SUCCESS in 22m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/051dede1976d41498053d3bf048d58e4 : SUCCESS in 25m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18fc0a5efc1648298ac3392df8db960f : SUCCESS in 29m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e46411a910f544fba1077a46ac44b1b5 : SUCCESS in 17m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d15b51cc50f4b24b9d8ab3bdb117213 : SUCCESS in 18m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/076b18be9ec04e338b4025659f56577f : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7e03565aa01149578ed299dad1295bfa : SUCCESS in 17m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a82bb56bde9f4adc8c7e968fa0e95bd2 : SUCCESS in 18m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2447baaed4754340b7259905e9be8498 : SUCCESS in 19m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a74e811336604d9eab83185e7ea8ce85 : SUCCESS in 20m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6369fc474e5143c69022e0a7e54840b3 : SUCCESS in 49m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/197ea03b5fca4592a6c7c6f26ac27d11 : SUCCESS in 52m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28e8d63707b44fbb9cb2c7bf7c36c397 : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e103647d8f324f60b300af27b17016c8 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd0a252f8f1c430caa87f4b1d867bfee : SUCCESS in 23m 20s","accounts_in_message":[],"_revision_number":1},{"id":"da56b23bfa2788f9faf528e4ea88ba6ad917ff25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 09:28:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c80545c4dad94c41b5eb49336d38bf9d : SUCCESS in 58m 38s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/099fa73110d34b7aa4a702ef2b367d0f : SUCCESS in 47m 19s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/f5bfc3e9e5ac43548d7b825da79dc1b4 : FAILURE in 1h 59m 03s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/510b5647670c4cbeab2b414793aac204 : SUCCESS in 1h 02m 42s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1dd28d29f9a040a28069c4fc4fe9d3b8 : SUCCESS in 1h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1df4c37f3ad0d05d381ca6f8cf34ed5c9fbf62ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-30 02:13:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3457ba7d19c5fff8c6266195d9686ab91a17ad5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 02:14:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41738b01215919ef70a463230db1cf3f9f33a0e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 03:31:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77fb28d299bc4e1d90bb2b457e06d4a4 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7a461c7323449849ce4f7db62c20315 : SUCCESS in 6m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/d20ee9cccbf248d5b5f3f8ba8925fa1d : SUCCESS in 1h 16m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c1874709aebf4d40978eb797156c6453 : SUCCESS in 45m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f311161c9478445784dd3700ac70d252 : POST_FAILURE in 8m 32s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ce5db7ea3e8b42399427bcf03a52e682 : POST_FAILURE in 10m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e70991e838a44dcaa4e68769d3ffdfe : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c4fed5544714a9eb4c0245ed458dada : SUCCESS in 17m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a23f168146f4447b9b779d483f11b4d0 : SUCCESS in 18m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f8571c16244c6bb3d83f248aaa4062 : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/355a7db4475043b8b60c79ba9ca42fff : SUCCESS in 58m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de83a7184c4c422d9cb31adc3a7ce379 : SUCCESS in 21m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2a71f2e43d5d48aea3e6408bf1829848 : SUCCESS in 21m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd29c041d3514586866647a5db10e87d : SUCCESS in 22m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9242552d6736486faf8b95f5eb827448 : SUCCESS in 21m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/876a2cfadaa044b3a864a407548f0439 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/03e0df5deb5f421f81748859729bfcd5 : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7d8cc21d81104aa2a823446d4601d7af : SUCCESS in 7m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/862ecf97c7744354ada4ddeb55c7eec5 : SUCCESS in 48m 52s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6fdffd339f8a44ab91fa215ccf0cf661 : SUCCESS in 47m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6cc2a133a16a4ea4bf7d938bc544ed51 : SUCCESS in 38m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6a7e62d46a34e4b90c61587fc083509 : SUCCESS in 41m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c63f25126178476894401d108f9c7d9b : SUCCESS in 14m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52de9752e4514ccdbcc338f35c2823d0 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d340a6d1c8fe41588416dd7314b83193 : SUCCESS in 25m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/178d3033eee24700b9e219154521a0a2 : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":1},{"id":"6096f3de19d6452cdd285927f64ee78dac96763c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 03:31:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4bb8aa1b708623ef9afe6392db50d015c2b6c1ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 03:33:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b08dce575fb401ba3b62fb4abd3c80c : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fa2523a963f4e2fa76099d0ac8d88d0 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e19b44abcb2745fdb8f625d6ffa50f7f : FAILURE in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/25578187408e41aa8b807211cb4608b5 : FAILURE in 46s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7520aada3c042fe209e2e4eb9414c0c3f8dfd027","revisions":{"7520aada3c042fe209e2e4eb9414c0c3f8dfd027":{"kind":"REWORK","_number":1,"created":"2021-09-24 07:23:50.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/26/810826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/810826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/810826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/810826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/810826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/810826/1"}}},"commit":{"parents":[{"commit":"29c314ea7ee63e66522dbaa6424c07bf02fbc803","subject":"Update .gitreview for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29c314ea7ee63e66522dbaa6424c07bf02fbc803"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-24 07:23:49.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-24 07:23:49.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","message":"Update TOX_CONSTRAINTS_FILE for stable/xena\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/xena branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ib67d08cbde3f952ae40a549d8458bf1d2dd99c7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7520aada3c042fe209e2e4eb9414c0c3f8dfd027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7520aada3c042fe209e2e4eb9414c0c3f8dfd027"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
