)]}'
{"id":"openstack%2Fswift~811112","triplet_id":"openstack%2Fswift~master~I1f46d53c9f7efbb666ed244af8b88994ea9380ac","project":"openstack/swift","branch":"master","topic":"dont_merge_epoch_via_replicator","hashtags":[],"change_id":"I1f46d53c9f7efbb666ed244af8b88994ea9380ac","subject":"Ignore remote own_sr with None epoch when local has an epoch","status":"ABANDONED","created":"2021-09-27 11:12:15.000000000","updated":"2021-09-28 10:18:58.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5238a6cdee7f17ff34a216d8083145a7ac4ccb5d","_number":811112,"virtual_id_number":811112,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-27 12:29:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-09-27 15:46:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-27 12:29:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 15:46:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"7a64fbac9449864bfa3c5ec880886203ad4c95fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-09-27 11:12:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"449fe101b784338f5e2fdcc7b13ff098ef598a93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-09-27 11:28:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9a37e3d68e860bec6275c8ffb435f667743d538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 12:29:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2be54e603348908d8ed8a31a53d7fe : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/8cc6671f883e43b7b8767add167009aa : SUCCESS in 49m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c27b5c0428b44ca4912e57d392c7a809 : SUCCESS in 41m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0955dc040f1141c39fd77e1e50a2e305 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/932f21ec109547f8aaec81f24577cbcd : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/54cb2ca6ac19447cb30d814ec98b4345 : SUCCESS in 17m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d49f9891cd3f4ad6b51d1d31288d2d80 : SUCCESS in 16m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/013da20cad584b54881a38bcba352497 : SUCCESS in 17m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/06ccc27cd06140468070dd5ec6edcf4b : SUCCESS in 17m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/482dfcea428b4d4b8fee79a2da717aaf : SUCCESS in 17m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4745ff7caf9b48e5a864c94fee3b34b4 : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/395e825b53cd44bda523ffcedcb82cf5 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff5e325700fa42b19395d495bc1b1c90 : SUCCESS in 18m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/62e37058ac8c47a88ab5cf885806dd73 : SUCCESS in 20m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8f86e81a9c3e4ee28ccfa711903d0fa4 : SUCCESS in 26m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/825560e9c90647c0b334e7ec5906b921 : SUCCESS in 19m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/169e64c21eda4cd1bd9ec0721bcda3c2 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a75d358c0e1b414fb53432d5a3a50257 : SUCCESS in 23m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d07c3c20c454647a0ec6d9fa7f5eb74 : SUCCESS in 53m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d582a9d42004455bb5a82e7e70c7197e : SUCCESS in 57m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57883cb823884b6789e5e9d48b16b7f5 : SUCCESS in 42m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/43462bae3b4b4d86b58e6dce8fd6ed5a : SUCCESS in 37m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d599120f5e5425cbb41cc772ed39e55 : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0fa4f77404c4a078d31c6a3d6dd2645 : SUCCESS in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ea2fd1500f34db390b0f8bf08309b72 : SUCCESS in 23m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a3d05c108d2413ba9da1024dcb4d245 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":2},{"id":"afd97bc0062e7d8d9f8fac54cbfb2af7e56a32be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 13:01:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6fed54fd9b747abb6226cce3bfaeb68 : SUCCESS in 35m 35s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c29d79f3c20494196a565eeef9fd5ab : SUCCESS in 35m 00s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/80c4e457bc9142059c13ddd64209626f : SUCCESS in 1h 24m 52s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/789f184b2a334335bdaadc24cafc4672 : SUCCESS in 36m 16s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/adf219ad7a5941cbb71907899e6d4d19 : SUCCESS in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0cec227fce050ae68d0db49e409ae00b01e33f08","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-09-27 15:46:02.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nmuch want","accounts_in_message":[],"_revision_number":2},{"id":"5238a6cdee7f17ff34a216d8083145a7ac4ccb5d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-09-28 10:18:58.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/809969","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"210fbca0cf4701ac147cb70c5e12f1fd640b807d","revisions":{"4b7acaad5745d59268c11817b81e7a58910b29a4":{"kind":"REWORK","_number":1,"created":"2021-09-27 11:12:15.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/811112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/811112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/811112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/811112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/811112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/811112/1"}}},"commit":{"parents":[{"commit":"ef0fbbbe0f3555ec4d22fbf6b79bb686b6adccbd","subject":"ignore osr with None epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef0fbbbe0f3555ec4d22fbf6b79bb686b6adccbd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-09-27 11:11:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-09-27 11:11:27.000000000","tz":60},"subject":"Ignore remote own_sr with None epoch when local has an epoch","message":"Ignore remote own_sr with None epoch when local has an epoch\n\nChange-Id: I1f46d53c9f7efbb666ed244af8b88994ea9380ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b7acaad5745d59268c11817b81e7a58910b29a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b7acaad5745d59268c11817b81e7a58910b29a4"}]},"branch":"refs/heads/master"},"210fbca0cf4701ac147cb70c5e12f1fd640b807d":{"kind":"REWORK","_number":2,"created":"2021-09-27 11:28:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/811112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/811112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/811112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/811112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/811112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/811112/2"}}},"commit":{"parents":[{"commit":"ef0fbbbe0f3555ec4d22fbf6b79bb686b6adccbd","subject":"ignore osr with None epoch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef0fbbbe0f3555ec4d22fbf6b79bb686b6adccbd"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-09-27 11:11:27.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-09-27 11:28:07.000000000","tz":60},"subject":"Ignore remote own_sr with None epoch when local has an epoch","message":"Ignore remote own_sr with None epoch when local has an epoch\n\nChange-Id: I1f46d53c9f7efbb666ed244af8b88994ea9380ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/210fbca0cf4701ac147cb70c5e12f1fd640b807d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/210fbca0cf4701ac147cb70c5e12f1fd640b807d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
