)]}'
{"id":"openstack%2Fswift~811247","triplet_id":"openstack%2Fswift~master~I150cf53b07e7d2d8de1d6e8c1fb08c07b9afe842","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I150cf53b07e7d2d8de1d6e8c1fb08c07b9afe842","subject":"Bug: fix s3api multipart parts listings","status":"MERGED","created":"2021-09-28 00:23:27.000000000","updated":"2021-10-03 23:48:00.000000000","submitted":"2021-10-03 23:46:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"811247","meta_rev_id":"31a398288ad51241dd5c7e45fb5a000af18dc20d","_number":811247,"virtual_id_number":811247,"owner":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-03 23:46:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-03 22:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-03 22:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-28 01:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-28 04:06:13.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-10-03 22:50:07.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"59606a351cba68c199291dfcc4b0b6f8c819c113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-09-28 00:23:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"372f17cbf8c706c50d26f599502dd178b57f7896","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-09-28 00:27:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b1849e120a7fefaa5649cb95c415e00e0254a708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 01:30:43.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b32c9b3a6648198ad3600e7c9b2593 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/b60113008b4a4e18a153ca1aaf48ec1f : FAILURE in 26m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f69fbded6c940b0a655670a25f95b01 : FAILURE in 25m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29946b6522d64661a8b50a8b332a4bbb : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/530fe4082dcd4b3ea164c9b5c15116eb : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1190bee381644609aad4dc0dfade5214 : FAILURE in 20m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af901936bcdf4152a72c536278e3561e : FAILURE in 20m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3c17cd650df04d3e81855c6da518c862 : FAILURE in 21m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/564b0e0529a84213acf97a7983f88fbe : FAILURE in 20m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/def1b92d6d554f2aa640bfa8a0e4d2de : FAILURE in 20m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7fdb19227ab40df929862e9baf31606 : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ad1273a35a14b28a37487e021b2e3a4 : SUCCESS in 22m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0c6b92fa74c4218af9e6e94901b3e94 : SUCCESS in 22m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/965b7451b5684fe1b3aa33fee9333c3f : SUCCESS in 21m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/71b3697dd0a3482baa743b0103c1a243 : SUCCESS in 19m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b20f2b8c29ae476985c8de3e2e2aeb41 : SUCCESS in 23m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be7c8ecf13524dd6a3e7376d44123c11 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/39ae4eeac7634af1975cc5fbad5a6b14 : SUCCESS in 22m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf24c44487694af98aff2bac7f1f2a62 : SUCCESS in 51m 04s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/77aa449c9fea45ffbf49174559d328c0 : SUCCESS in 1h 02m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b86d5dea1b134d049160e7bff6f42d7c : SUCCESS in 41m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd88d5e425be4692a11d1aa11c8bef68 : SUCCESS in 42m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fcb658e4b5f4e9ca2bed59cf5672c5a : FAILURE in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf0089d5c364dd7b9510b4e54a5ded3 : FAILURE in 6m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/127e32d015594b0cbfcc4bf00d7df75e : SUCCESS in 30m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0af3df444e86451488cab21c04252402 : FAILURE in 28m 49s","accounts_in_message":[],"_revision_number":2},{"id":"0a2edf57f5e71cf1c48843e21acaf419d6de001c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 01:56:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/703d2e9bc4c34ee49a0c595fae3acd25 : FAILURE in 46m 40s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a5a79c4d80e4d4587c4c3734ba09818 : FAILURE in 32m 05s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/e673da5732ad42cab391a17e8949f220 : SUCCESS in 1h 22m 37s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/948b361f086f4d329978b732a70116db : SUCCESS in 34m 38s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbbce81d8a36423cbd9adbef1be69ad1 : SUCCESS in 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b156a27cdad5dbc9dc86150fe33e325dcba5791","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-28 04:06:13.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2a8885ae3e2706fe5e68c604d519904c96cb32b0","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-09-28 15:52:22.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nThanks for checking this out!","accounts_in_message":[],"_revision_number":2},{"id":"883f590c09582643fff32e21a30528afba02ccab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-09-29 01:04:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"643b0198a04996e1de00e5c1c3e772285a97b10b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 03:36:52.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbbd456e0df747638905e96ccbfaa61a : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/ab07f74765d8423cadea6674ef595d43 : SUCCESS in 1h 03m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68975d12c6414a7d84b4a487c6b028e9 : SUCCESS in 46m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9185942f12ae4fd5855ef3596889b08e : SUCCESS in 9m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d1d61a79243f4927953e56f4fb5722ca : SUCCESS in 10m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3bb8e14aaa684c4ab416f496a75525dd : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/39faa13b518e412bb2cc6ac71d9cbea8 : SUCCESS in 20m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1af8fc7245449e190f01c5981f52762 : SUCCESS in 18m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1c38e598c7846b59949ee4e9bc42c95 : SUCCESS in 17m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/17d428184f4d411ea4f23c82b58d1d63 : SUCCESS in 19m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8eae03233adf424eb062f3b0e65835e4 : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/80db640dcd7f478e8e843daa55710607 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/927d0eddc27f4e4a8b62b6875237e0f4 : SUCCESS in 21m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/27c426712e1b4da0858521350ce14682 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/00a18fa20cc24e5eab9786d0a81d98d3 : SUCCESS in 20m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/929f33a1df5b499e8a64ce788b4948f7 : SUCCESS in 19m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d5cb52e862c04fdab78ace71bbc5f6f8 : SUCCESS in 8m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5354a4147f474ebea6e796f9317d4a47 : SUCCESS in 23m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9847871fec5242a5893fc0f079fc480b : SUCCESS in 51m 53s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/68fe9e8d688f4edb8605659d9f501eb7 : SUCCESS in 50m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce1a3e5c55b846b09804b12b305541c0 : SUCCESS in 40m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/baa2a6ece20e4ba48946e113f3a2c2f6 : SUCCESS in 42m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/389fb0fed1d044c581140a3ddea2f104 : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44360b6f0dba4aa99f09d93fc49407dd : FAILURE in 8m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9df0632957c24ad8847d1679cf2380d7 : SUCCESS in 31m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f283782078a4f7c8f1a0c84b33c116a : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":3},{"id":"39a06ef0a7c12dd6f25c794489ffa306a4fbc886","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-09-29 03:59:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"951d331b0a017b4355b2815384d763634e1b2ea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 05:10:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ee65b8d9ddb45ee9113e164d4e65f6c : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b0ca2e0a13b04abfb6e1781dd662cc15 : SUCCESS in 1h 04m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ca106c6b135494f8fbba14d748651f9 : SUCCESS in 42m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de9d02510cb1477db7e1516cc4dcbda4 : SUCCESS in 8m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/95eafb6693bd4e43b21fcf4176f94541 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c8e9255538e445ea0e37a747be00163 : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07030950bfb3485c8149ba2f3a301977 : SUCCESS in 24m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e7df48a4cf2a4022a12d2169eb4439fe : SUCCESS in 20m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5735d8482d45436281d4a35ff6c3ee31 : SUCCESS in 16m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d674254468e14c1c8f05c536fb86fdd5 : SUCCESS in 19m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1e9cfcd335434d82bebb74c95d60f8c8 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ef7be8eda774e8d9296faaeaf4626ec : SUCCESS in 17m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac8713003edd492a9c581519cb5ca45f : SUCCESS in 21m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eff6c48b51f64e64840bd5ff20ceebf1 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eea864eb614240229c8a2ee4273f9804 : SUCCESS in 19m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/41b0e9983ccf44e1b5102663d2f3b777 : SUCCESS in 20m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ab2a06b5d92403faa523702beee6c18 : SUCCESS in 6m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0adc95ad646e46619e043f97e3abfd65 : SUCCESS in 23m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b50c5451d110400290613899919e012d : SUCCESS in 52m 44s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/98586587074f432a96e8261205c13f38 : SUCCESS in 58m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/681129b10e244a0f9f3e2bebe2dfe62e : SUCCESS in 49m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7630a6093bc04e618df3f84cfecffa41 : SUCCESS in 37m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27f859356dce4420acb2bfee4979b1a1 : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ad4a7836d2410096f63b554d9af189 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65e0523030ca4d289d62cf57ef2b541d : SUCCESS in 29m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d05dd5b538f48be97a9f0c4fd00dac2 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":4},{"id":"6c697c758e50df528c6a72571e73afcf9768c10d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 05:57:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4b22f77e3224483a80673c21854cb58 : SUCCESS in 33m 12s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47f30c725ee647e2ba9c162e27bc890e : SUCCESS in 31m 17s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/9cf5b65d5ed248e8a368027c1bd101cf : FAILURE in 1h 55m 15s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3eea3e795dd24b4ab22a9da4b60947d0 : SUCCESS in 59m 19s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5deb59b25324a3bbc03c471c98ae0dc : SUCCESS in 59m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c078fff12727916f0dcfef8d02efdfec4bcda5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-03 22:50:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"265818d5bf2ad2f518d7abddf5b8eb6a58f6b8f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 22:50:43.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9749b4f2ef37cc06c2aeb47a8d8e0f5b1efc87fc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-03 22:51:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1418edd8e2533c049519e813c3ba6cfdcd45ec26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 23:46:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97930fdb99554af8b493550175e40f08 : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/6dd6be8e8b4945ee96e2a04ac55ded70 : SUCCESS in 48m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdc94ea10a464262bdecb5bfd37fab71 : SUCCESS in 36m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ff20945cea294915993c2b736403b50b : SUCCESS in 9m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/93be4ef2f1cf4af9901b783695cc6fca : SUCCESS in 11m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbf5235d06eb43199dbf3f08d3dd6bd7 : SUCCESS in 16m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/523ec3b998b64e03936936fb2d446911 : SUCCESS in 15m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f76790a658804657a62688be7bd8a659 : SUCCESS in 15m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f3e5dd889fd94c1b9d96656dcd13b6e1 : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53bdb9d6e26d47afbb952082ea809d56 : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be9d76f2dfe14b2b927f667555fec877 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2c8cbcd8e4b2418dbd52cc660c616e06 : SUCCESS in 17m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d9b82c01a23f4c75a9bd823b80dc7caf : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/57637aa93f6f4195992942df90d4c5eb : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/777ebadff3b0474289d1a47238f42a29 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fe3fa8e798b84b0b9ce481b3966a7d99 : SUCCESS in 18m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c710ebe6ab2441ff8b15fdc722c80664 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/14f67e13616c4187a16c401394f7a88a : SUCCESS in 49m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9ee313ae7f82433db4d4d73c5bf1ab49 : SUCCESS in 51m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/56611426588e4ed28352152202c054e7 : SUCCESS in 46m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b1e25c34e341490fadae372062435acf : SUCCESS in 38m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f9ae5c7e4d649a1b2f1e4712c05aadf : SUCCESS in 30m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4ff3d6908f470b81d93f7898042d34 : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3cfa30f7d2b467c93bbc83ab114bc72 : SUCCESS in 26m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83290715728b4c11a91b83d98919432e : SUCCESS in 38m 56s","accounts_in_message":[],"_revision_number":4},{"id":"873be528fbc41b545d38b1d26f192e79a7c869cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 23:46:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"31a398288ad51241dd5c7e45fb5a000af18dc20d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 23:48:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f14c6c7d534985a0f93e3daeed09f9 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/68aad8a0be3241258022d3512c6e71ce : SUCCESS in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/72d2d6c9d56f4c0db7fe9e7a74a15de1 : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"47749cd0e53e5178972a699f47bdcedbde45b544","revisions":{"d1f5b9f1e9139db7db6ecc6752209f74702b57cb":{"kind":"REWORK","_number":1,"created":"2021-09-28 00:23:27.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/47/811247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/811247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/811247/1"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-24 19:05:04.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-28 00:23:12.000000000","tz":-420},"subject":"Bug: fix s3api multipart parts listings","message":"Bug: fix s3api multipart parts listings\n\ns3api returns multipart parts listings out of order and possibly\nmissing. For example, if there are 2000 parts, the first 12 parts\nreturned by s3api currently will be: 1, 10-19, 100. Then after part\n189, the follwing part is 1000.\n\nThe change fixes this behavior by internally listing all of the parts\n(with default settings, this should be 1 listing request, as the 10000\nparts limit matches the Swift listing limit). After that, the parts are\nsorted and delimited/marker settings are applied to craft the response\nfor the client.\n\nChange-Id: I150cf53b07e7d2d8de1d6e8c1fb08c07b9afe842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1f5b9f1e9139db7db6ecc6752209f74702b57cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1f5b9f1e9139db7db6ecc6752209f74702b57cb"}]},"branch":"refs/heads/master"},"c2d1783e22c61ab645754598b918ba4b538c7527":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-28 00:27:18.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/47/811247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/811247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/811247/2"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-24 19:05:04.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-28 00:26:34.000000000","tz":-420},"subject":"Bug: fix s3api multipart parts listings","message":"Bug: fix s3api multipart parts listings\n\ns3api returns multipart parts listings out of order and possibly\nmissing. For example, if there are 2000 parts, the first 12 parts\nreturned by s3api currently will be: 1, 10-19, 100. Then after part\n199, the following part is 1000, and so on.\n\nThe change fixes this behavior by internally listing all of the parts\n(with default settings, this should be 1 listing request, as the 10000\nparts limit matches the Swift listing limit). After that, the parts are\nsorted and delimited/marker settings are applied to craft the response\nfor the client.\n\nChange-Id: I150cf53b07e7d2d8de1d6e8c1fb08c07b9afe842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2d1783e22c61ab645754598b918ba4b538c7527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c2d1783e22c61ab645754598b918ba4b538c7527"}]},"branch":"refs/heads/master"},"d5011dfb9460a9e10059887cfc03dbc78457151e":{"kind":"REWORK","_number":3,"created":"2021-09-29 01:04:50.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/47/811247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/811247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/811247/3"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-24 19:05:04.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-29 01:04:28.000000000","tz":-420},"subject":"Bug: fix s3api multipart parts listings","message":"Bug: fix s3api multipart parts listings\n\ns3api returns multipart parts listings out of order and possibly\nmissing. For example, if there are 2000 parts, the first 12 parts\nreturned by s3api currently will be: 1, 10-19, 100. Then after part\n199, the following part is 1000, and so on.\n\nThe change fixes this behavior by internally listing all of the parts\n(with default settings, this should be 1 listing request, as the 10000\nparts limit matches the Swift listing limit). After that, the parts are\nsorted and delimited/marker settings are applied to craft the response\nfor the client.\n\nChange-Id: I150cf53b07e7d2d8de1d6e8c1fb08c07b9afe842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5011dfb9460a9e10059887cfc03dbc78457151e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5011dfb9460a9e10059887cfc03dbc78457151e"}]},"branch":"refs/heads/master"},"47749cd0e53e5178972a699f47bdcedbde45b544":{"kind":"REWORK","_number":4,"created":"2021-09-29 03:59:36.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/47/811247/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/811247/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/811247/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/811247/4"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-24 19:05:04.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-29 03:59:31.000000000","tz":-420},"subject":"Bug: fix s3api multipart parts listings","message":"Bug: fix s3api multipart parts listings\n\ns3api returns multipart parts listings out of order and possibly\nmissing. For example, if there are 2000 parts, the first 12 parts\nreturned by s3api currently will be: 1, 10-19, 100. Then after part\n199, the following part is 1000, and so on.\n\nThe change fixes this behavior by internally listing all of the parts\n(with default settings, this should be 1 listing request, as the 10000\nparts limit matches the Swift listing limit). After that, the parts are\nsorted and delimited/marker settings are applied to craft the response\nfor the client.\n\nChange-Id: I150cf53b07e7d2d8de1d6e8c1fb08c07b9afe842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47749cd0e53e5178972a699f47bdcedbde45b544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47749cd0e53e5178972a699f47bdcedbde45b544"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
