)]}'
{"id":"openstack%2Fswift~811387","triplet_id":"openstack%2Fswift~master~I309280892515dcfd196c511dfd2df0ba3bab2b74","project":"openstack/swift","branch":"master","topic":"p-smsr-enable-force-epoch","hashtags":[],"change_id":"I309280892515dcfd196c511dfd2df0ba3bab2b74","subject":"manage-shard-ranges: fix traceback if own sr missing epoch","status":"NEW","created":"2021-09-28 10:07:23.000000000","updated":"2021-09-28 16:51:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"247f96d39a667dd59635697f295c1f20a3db7dd2","_number":811387,"virtual_id_number":811387,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-28 11:05:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-1,"date":"2021-09-28 16:51:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-28 11:05:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee8020c64df02895bdf5a49ad5f78fdedf8f6d75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-09-28 10:07:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe79dedf41dbb78183c63bbea549b45da187d758","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-09-28 10:11:17.000000000","message":"Patch Set 1:\n\nI don\u0027t think this is the cause of what we are trying to pin down with [1], but it is curious that at some point the possibility of a SHARDING own sr with epoch\u003dNone was considered.","accounts_in_message":[],"_revision_number":1},{"id":"19d7759b655276e6e6aa7c83070009350833ec2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 11:05:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/368353306e584310b30c5283127993ef : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/588cbca87f5448ac9538a1e5f06a42f0 : FAILURE in 29m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05e147dd20e04999a1359e0b776c585c : FAILURE in 27m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38f2eb83f0114d6790e78859c7000376 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/53eead1dfd03470c8a089567354af5ce : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d7ac13cc8f44d98ac61e653c4d590a1 : SUCCESS in 18m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1befa1ac47014fb9a524173fa8217540 : SUCCESS in 16m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b019430a1244b7ca1e45e3e698ad315 : SUCCESS in 16m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/86aba2e78c154412b359f3c2a8a25bc9 : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/588580d46cd640a281f116b6a9ffad58 : SUCCESS in 17m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8e122d319fcc4ff2824187c9e6e17463 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed28dc775fd147ab85ad64915b6cc594 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d4b1d1a4bb64eb3882e975f9ebd9ced : SUCCESS in 17m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0902e3ba9027405fb145bd4f6a652bae : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e2f7ed5182d44dfa665dbcee20b3250 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/61ac5f9c0f9442d4a5e0106f53606848 : SUCCESS in 19m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ccc5e52639e2414db2d31115f2c5ee27 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8128e390e3ca4c57b803c8a42c6630f5 : SUCCESS in 21m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3700683cb59c4707ae3f9522398ca07d : SUCCESS in 47m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/61f9e73bcb4c4df2bf50104ae1b53a00 : SUCCESS in 52m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46ba68f510fa49738acd9dcfb4983159 : SUCCESS in 37m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fe69f8764854fd18b6ce83caf314e7d : SUCCESS in 40m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce41c37cec5f4434a75104f29c8f2a1e : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c2b94d7292456fa5edbae6f574f61b : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6890908b401b4312938eb245ef4eb03e : SUCCESS in 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac1bb681492b4af3a1fb5209b0e36230 : FAILURE in 28m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3b779d6e55a625838ebc3fbdc0fb179f866979a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 11:56:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c24822acc9b43de96615575d3beb7d5 : SUCCESS in 37m 06s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88cf975549044437ade8fea34777d483 : SUCCESS in 32m 01s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/746bc57d44654d3781ca6dc570bc13f4 : SUCCESS in 1h 40m 00s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d17502f7643f42a6add03fa5ff4c9242 : SUCCESS in 1h 09m 05s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ce5fec1d35c4d71b2ab0f99175bc71a : SUCCESS in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"247f96d39a667dd59635697f295c1f20a3db7dd2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-09-28 16:51:44.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027m not sure if fixing the traceback by forcing an epoch is the right thing vs. turning this into a CLI error (because we don\u0027t *expect* this SHARDING state with no epoch condition to exist, do we??)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a21f296df9c9d7457546c0c44bbf45ed6c31686","revisions":{"5a21f296df9c9d7457546c0c44bbf45ed6c31686":{"kind":"REWORK","_number":1,"created":"2021-09-28 10:07:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/87/811387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/811387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/811387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/811387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/811387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/811387/1"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-09-28 10:02:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-09-28 10:02:07.000000000","tz":60},"subject":"manage-shard-ranges: fix traceback if own sr missing epoch","message":"manage-shard-ranges: fix traceback if own sr missing epoch\n\nWhen enabling sharding, the manage-shard-ranges CLI allows for a db to\nalready have an own shard range in SHARDING state. It also checks that\nthe already SHARDING own shard range has an epoch, and attempts to set\nit if not. However, the attempt to set the epoch will always fail\nbecause it is consitional on the state being updated, which it is not\nsince the state is already SHARDING. This results in a traceback when\nthe epoch (still None) is printed.\n\nIt\u0027s not clear that we should ever have an own shard range in SHARDING\nstate with no epoch, but if it does occur then the epoch chould be\nforced to be set.\n\nChange-Id: I309280892515dcfd196c511dfd2df0ba3bab2b74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a21f296df9c9d7457546c0c44bbf45ed6c31686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a21f296df9c9d7457546c0c44bbf45ed6c31686"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
