)]}'
{"id":"openstack%2Fswift~811388","triplet_id":"openstack%2Fswift~master~If837d4e55735b10a783c85d91f37fbea5e3baf1d","project":"openstack/swift","branch":"master","topic":"fix-non-ascii-object-name-copy","hashtags":[],"change_id":"If837d4e55735b10a783c85d91f37fbea5e3baf1d","subject":"s3api: fix the copy of non-ASCII objects","status":"MERGED","created":"2021-09-28 10:07:59.000000000","updated":"2021-09-30 06:01:40.000000000","submitted":"2021-09-30 05:58:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"811388-fix-non-ascii-object-name-copy","meta_rev_id":"d523e7285dac9a29fc8c654907de522ef1c35f77","_number":811388,"virtual_id_number":811388,"owner":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 05:58:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-30 04:53:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-30 04:53:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-28 11:07:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 02:05:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"8443f8ba6a80714cf456a395b83a425325c062d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"date":"2021-09-28 10:07:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6abbd77e0dcb4fb167d7a82491c01a52775c40f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 11:07:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed1668389f324c00814be02c5606ba91 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/b60207cb29984429acd47663768aacff : FAILURE in 20m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/427e088e05e14874808e46495ef23cba : FAILURE in 27m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5910a7a6e8804205bafe216f7d38ed8a : SUCCESS in 8m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6cb9fc901f5e45149ab93a4f55b6986e : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33394a5bccb742e4ab4a9cfd9a2fd64f : SUCCESS in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9b3aa9cd7f8413b96f12c3f3068d9ae : SUCCESS in 17m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bf145e2c180f419180902001a05191fc : SUCCESS in 16m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1af0966e8a1430a83aa6c51dcc1d798 : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce0924cc16a498d836853261a967c9d : SUCCESS in 18m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/817f39319d4242df8767862039ffe82f : FAILURE in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af29f62a35824b93a32f6f2b8f7b9a2f : FAILURE in 20m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d5a641d167c4bebb0cb365c67bf4c73 : FAILURE in 20m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/425ad21c96ab4ff78920ed80dd916cdf : SUCCESS in 19m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82c250a5b1c64cf3b536ba26d3f4ff9a : SUCCESS in 29m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d37e9c3f90a542509623318d0edc875f : SUCCESS in 17m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3f4f372235c2450eb0fd4572007a7712 : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2c9f980caaaf4858a5b1d105de7bf40c : SUCCESS in 24m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51255bb2c28f458d8b5f2c4639557d0a : SUCCESS in 51m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c54f0174877a4b588594d65bf189f41b : SUCCESS in 56m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/97a758319d8249e39aa933cceed0d76c : SUCCESS in 38m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e527485b7e3d46a79092b1a7631e9e7c : SUCCESS in 39m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04816de5c79a45c59b589f7c6f94beac : SUCCESS in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1add3bbb624b8ba8df9c6c704b864a : SUCCESS in 6m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e35bc3e4667a4e3aa144f51d75040d9a : SUCCESS in 22m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca546e665b4b422fb76d0fbae961a6ea : FAILURE in 27m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1862ac2f0efc4e8ecd199fb0f2c30aa5ef88e230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"date":"2021-09-28 11:25:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32e5fabdb9c34b97682b48c27a63391e32451fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 12:23:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ea9a60c0fc4d3799f68cba219935db : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/8225f9914fa04a59b227dc6d1af0fc01 : FAILURE in 31m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38ac84a2819d451380a2462c54687910 : FAILURE in 31m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ef8c1b7790a4f0399ed85b0b62c5820 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b53d96c4799a4258a9491675d461fc7e : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d117e59aaba94daa99d89bbffa01b1db : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df4b4f0772864db9a311881e7f79e67c : SUCCESS in 16m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e7a077175f39410588010e0c6dfa2312 : SUCCESS in 18m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aac49ecfbff649c087e36ce576cb2c4c : SUCCESS in 16m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb73c8f00dc84c76a026aed92c19a387 : SUCCESS in 18m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f52068877a2148f391adc29b5cad4049 : FAILURE in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b1db83f1dc34931ae77122f599d5ade : FAILURE in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2509f5c13590484180f430d3aa79f66f : FAILURE in 18m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75a040f2413b46c38cce9ff307a28113 : SUCCESS in 20m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8b98dc565f024993a61345b7533c502d : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/51cafec6d2544105b709974b3333bd98 : SUCCESS in 17m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dffd4d9410624b1aacee42a30921dc5b : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/273a635641c74a9faffe56b2bc87f6d3 : SUCCESS in 20m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ae1808522d3d4cf7a4dac56374b18673 : SUCCESS in 47m 46s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1ba1bb87cbb84665971e6b7a98e6ceb4 : SUCCESS in 53m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/202cee8aab9a43c9a1a0b172f77c137f : SUCCESS in 40m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/890ce6633ee14a91939e17ef260edda3 : SUCCESS in 38m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/822a71d352ed430ca4fc24094934dc41 : SUCCESS in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d44d0565762e4b03a93ee045f95207b8 : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b024ce82d1fc4e21a66bd43860d53d78 : SUCCESS in 22m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaf7203f2b364da2b71d95a34b054a00 : FAILURE in 28m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a7e824e113e6a332e718549acd90500ab36ad2a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 13:17:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4515615f48d48418f97261b245895af : SUCCESS in 1h 04m 06s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29d50cabae174ca88612946999af2820 : SUCCESS in 47m 47s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/07f7a40f92a64f008eeadfe84ee6108d : FAILURE in 1h 45m 59s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/895a6b1334214b1fa53a32b2e0dfbad5 : SUCCESS in 59m 26s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/862feba6645943caa81a9351aeba8fac : SUCCESS in 1h 00m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65b63d6b224591e1e24ccb69aa93e465483855c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-30 02:05:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"54096e627fd6e68cb4e5096321edd4b4190691ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 03:08:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f581a4f5df0460f91f9ad3aca2afaa1 : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/e42888f7c912487a922a7c811d8a8807 : SUCCESS in 56m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac970308ee7e4a828c49d08b19892c31 : SUCCESS in 45m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/097bbc4fa62d4d0ca8ec8dd2d530e995 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/11901117d3e74151a05c65e6b04ae668 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6ade0a8c22a3481e908f0e69e8997aa6 : SUCCESS in 18m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ae381a362d645df80806a33d0ab2065 : SUCCESS in 18m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/455492dd0eb148cbb8fdc61a2e15d92e : SUCCESS in 18m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d93898673dbd4ce4b78e0328032221f9 : SUCCESS in 16m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/13a70ffb4f2a4d509c450772eae43478 : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b770f651f40414398b5688cbb5b88d9 : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e7ebd82009e447fa987cb0eed451690 : SUCCESS in 19m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d1cce84621394e6685f64f4567a2dab2 : SUCCESS in 17m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b322e2eb0e4c49dd8524009c3d4d3db1 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ecf91cb38a2b4541abfaf2243e40401f : SUCCESS in 20m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3c46d7c1f094622a7f3a3a62c5b5683 : SUCCESS in 19m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/722bde0bcd4c4b2481eb72ef7bf05dc0 : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cdaa15d275bc4e01864a05fd0aa6bbe3 : SUCCESS in 19m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f43f3d179ec1478f8c5329d1cd9dfa79 : SUCCESS in 48m 46s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8e7737ba895345719907036a755901bb : SUCCESS in 51m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1d6d93897584510bb89ff14b7c0c6fe : SUCCESS in 37m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79558eaa01cc4749b43dd70951af602b : SUCCESS in 38m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3266a203e99742198cd710abddb556a2 : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e404e8868c1846ea94db16025f92a24e : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b70dc3774fef43d69104775037c3f1a1 : SUCCESS in 22m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64295dae3bdb435d92b85ebb8adff703 : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":3},{"id":"32cca311b1ed2aa6602422001145e7e9f28cb400","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 04:09:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c4a03722bb5417da4361acf3d858af8 : SUCCESS in 56m 54s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6996389b2da147e188be44f387349717 : SUCCESS in 48m 57s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/1307123ea4184022915cc208c23b285d : TIMED_OUT in 2h 01m 23s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f21dd3bdacec43d68d40536c56f28c34 : SUCCESS in 29m 57s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b6693028e7241d5b72886da813084a5 : SUCCESS in 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a83f05e01093a83d00a42e5fd6a03203b445f1d8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-30 04:53:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThanks for this!","accounts_in_message":[],"_revision_number":3},{"id":"bd5f1ce17a0cf920efb238932f980188a66c78c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 04:53:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"863ee2889e2994af5213d3b6c4934f7c1c0470aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 05:58:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68b5b75757e5434a9362cd02121cd789 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/a6cafee2fdee4b448cdd7e4f6a89e2a2 : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aaba61c54b3c4df89a610e28c94f2217 : SUCCESS in 51m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f996264caf5c44098e2413a9f6ace883 : SUCCESS in 11m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8410c8a022f14af183ee378eb4288c03 : SUCCESS in 10m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b680fe8f751e40f79866c45f3e21b0a3 : SUCCESS in 19m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14ea738d2e1845e3b60ca9125524d26e : SUCCESS in 17m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e5581816ad8458e9bf5aa4e418dd1c7 : SUCCESS in 17m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1c10b3a1b68842239ae6cdee87e3d0c1 : SUCCESS in 17m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efbfdeb348d541fb9785ae2cee34bec7 : SUCCESS in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4df6d3cccc5c4fc997f587f459864eab : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c8047bb23f54ca7b5ab40a95526587a : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40ffebb2f3d54df2b27af532c6fe944e : SUCCESS in 20m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fc50750a2158489e8e72e67e9efae186 : SUCCESS in 21m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c82c98da2e334a0d8ede7fb56df4d49a : SUCCESS in 20m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1133b45c67c0417486fa2e2bbc1a94ec : SUCCESS in 19m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aaae746bd8864c21ad200d5ca6b8ff3a : SUCCESS in 6m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c65c82e9038b419997a902b9dc53807b : SUCCESS in 54m 10s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/61ebd07a96304366a62eff948e978767 : SUCCESS in 52m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/470a747d81314fa3bb2e4f1df1d5ac33 : SUCCESS in 40m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/14ab571c3ebb4771a3eddba460c55a51 : SUCCESS in 38m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beca433af3914033a84327219b902895 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc023c1ce9344e4a49e2d307eb75ab8 : SUCCESS in 6m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94285d73e6564f0ab64f98ff144b61ba : SUCCESS in 34m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2be64d6d7623400c9f38794a093ab9a9 : SUCCESS in 47m 32s","accounts_in_message":[],"_revision_number":3},{"id":"9d6c287085e0b48975b62ac0c0e96582aeff3861","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 05:58:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d523e7285dac9a29fc8c654907de522ef1c35f77","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 06:01:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d44f729e5122441b90753d3c55ead850 : SUCCESS in 1m 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/73d318da8e3d4b2fb76c5f5781f8599f : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/978d071c971c49be856eeb12ca676601 : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c15818f1e6f77119b57d2646c291d694cebd2698","revisions":{"d79e0d9765f5660b7ce9ea3bc436c3b700801d2b":{"kind":"REWORK","_number":1,"created":"2021-09-28 10:07:59.000000000","uploader":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"ref":"refs/changes/88/811388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/811388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/811388/1"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","date":"2021-09-27 16:43:42.000000000","tz":120},"committer":{"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","date":"2021-09-28 10:00:58.000000000","tz":120},"subject":"s3api: fix the copy of non-ASCII objects","message":"s3api: fix the copy of non-ASCII objects\n\nTrying to copy an object with non-ASCII characters in its name results\nin, depending on the pipeline:\n- an error code 412 because of a badly urlencoded path\n- an error code 500 \"TypeError: Expected a WSGI string\"\n\nThis commit fixes the problem by calling str_to_wsgi on the object name\nafter it has been urldecoded. We do not need to call this on the\ncontainer name because it is supposed to contain only ASCII characters.\n\nChange-Id: If837d4e55735b10a783c85d91f37fbea5e3baf1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d79e0d9765f5660b7ce9ea3bc436c3b700801d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d79e0d9765f5660b7ce9ea3bc436c3b700801d2b"}]},"branch":"refs/heads/master"},"d53bedacf836fb53a4dfd2a3ddc442f8a900b2c1":{"kind":"REWORK","_number":2,"created":"2021-09-28 11:25:33.000000000","uploader":{"_account_id":26274,"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","username":"fvennetier"},"ref":"refs/changes/88/811388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/811388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/811388/2"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","date":"2021-09-27 16:43:42.000000000","tz":120},"committer":{"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","date":"2021-09-28 11:20:43.000000000","tz":120},"subject":"s3api: fix the copy of non-ASCII objects","message":"s3api: fix the copy of non-ASCII objects\n\nTrying to copy an object with non-ASCII characters in its name results\nin, depending on the pipeline:\n- an error code 412 because of a badly urlencoded path\n- an error code 500 \"TypeError: Expected a WSGI string\"\n\nThis commit fixes the problem by calling str_to_wsgi on the object name\nafter it has been urldecoded. We do not need to call this on the\ncontainer name because it is supposed to contain only ASCII characters.\n\nChange-Id: If837d4e55735b10a783c85d91f37fbea5e3baf1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d53bedacf836fb53a4dfd2a3ddc442f8a900b2c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d53bedacf836fb53a4dfd2a3ddc442f8a900b2c1"}]},"branch":"refs/heads/master"},"c15818f1e6f77119b57d2646c291d694cebd2698":{"kind":"REWORK","_number":3,"created":"2021-09-30 02:05:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/88/811388/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/811388/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/811388/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/811388/3"}}},"commit":{"parents":[{"commit":"029e57679c639edda87816482685b8f48de67dd9","subject":"Merge \"Clean up extra spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/029e57679c639edda87816482685b8f48de67dd9"}]}],"author":{"name":"Florent Vennetier","email":"florent.vennetier@ovhcloud.com","date":"2021-09-27 16:43:42.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-09-30 02:05:14.000000000","tz":-420},"subject":"s3api: fix the copy of non-ASCII objects","message":"s3api: fix the copy of non-ASCII objects\n\nTrying to copy an object with non-ASCII characters in its name results\nin, depending on the pipeline:\n- an error code 412 because of a badly urlencoded path\n- an error code 500 \"TypeError: Expected a WSGI string\"\n\nThis commit fixes the problem by calling str_to_wsgi on the object name\nafter it has been urldecoded. We do not need to call this on the\ncontainer name because it is supposed to contain only ASCII characters.\n\nChange-Id: If837d4e55735b10a783c85d91f37fbea5e3baf1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c15818f1e6f77119b57d2646c291d694cebd2698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c15818f1e6f77119b57d2646c291d694cebd2698"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
