)]}'
{"id":"openstack%2Fswift~811876","triplet_id":"openstack%2Fswift~master~I5357a095110e8e4889c0468c154611209c6e8c07","project":"openstack/swift","branch":"master","topic":"nomodulo","hashtags":[],"change_id":"I5357a095110e8e4889c0468c154611209c6e8c07","subject":"Get rid of port to node assumptions and their modulo kludges","status":"MERGED","created":"2021-09-30 05:42:49.000000000","updated":"2021-10-01 17:59:06.000000000","submitted":"2021-10-01 17:56:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"811876-nomodulo","meta_rev_id":"9b72753eb35420ee32878ba06982cbc768d1be5b","_number":811876,"virtual_id_number":811876,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-01 17:56:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-30 15:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-01 15:20:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-30 06:40:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-30 15:46:30.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"947e74f008b01e399973a25cd6074ffb6d881b72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-09-30 05:42:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0939aa1fb0e0c661561f75fb5e0cdbe31edf18ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 06:40:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/622773e71fde454f9d46069c1bf900e2 : SUCCESS in 9m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0d39883d660b49dbabdd1d12fa5457e5 : SUCCESS in 55m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/70bd5ac7a6834e85baba565d1da6f7d6 : SUCCESS in 53m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bea54e252447849868e11ae085f642 : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":1},{"id":"dbd9e0474d2ad437a6809c05b468981c2723e56d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 07:20:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59eae8c6d49147f0b49d5a068c234e87 : SUCCESS in 44m 42s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdd22df60e37487e952dc165cd60a2d1 : SUCCESS in 33m 25s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/a00b2c4b1f8d4830b636894f5f3b0915 : SUCCESS in 1h 23m 01s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d5143a60cd94052aac80437d53a85e9 : SUCCESS in 1h 05m 46s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/836452385993445d9ba01b29f3251f2a : SUCCESS in 48m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6d6175d0c208aa62f874ac0c25b9653336018ba7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-09-30 15:46:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f55b6b4d377cf273f61d696053504e159ea6757e","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-09-30 18:16:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bdbf74130d8bb30b15bf33e1f09416799e5b80c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-09-30 18:17:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b7fd3dbf44041576116745e39ea3235c034ccb2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-01 15:20:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6736ddf66179bda0d361975f01dbd5bc5756941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 15:21:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0a71926eaf93bce77ebf229dd1cea13bb8a30a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:56:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f06ef9819a845bf89c69883d440a17e : SUCCESS in 12m 32s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f900a53a18fa4660b6108313d99f8db9 : SUCCESS in 19m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03202831df104881baeb3a23f40b7004 : SUCCESS in 18m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d630cb34aeba41bead0c8823a3732531 : SUCCESS in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/24750efb2bba41979b111ee4a54a61b2 : SUCCESS in 20m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8340669bc5aa4398800a6aaa3566b4dd : SUCCESS in 18m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/98dcaa6c846d4f5b96301a2950fe6fea : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d781f1d4fb4349dfacaa5164d68db249 : SUCCESS in 26m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8cfe3a596e7423da742cf6fc4c88917 : SUCCESS in 26m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/da3ff818b63449c0850bb5ff9ec48fc5 : SUCCESS in 22m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/34117ac3ef7e4c10bd481fa4b1ed3752 : SUCCESS in 17m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b8bbc9dd18e74da69119bd7961590dd7 : SUCCESS in 22m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db40ecd8379941fc8c87d31d6bcf50f7 : SUCCESS in 8m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d2c64aa1b4a14839991d5959c6721057 : SUCCESS in 55m 03s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3d1ca210b11949989feea513501dc2b2 : SUCCESS in 58m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e623cc975894d869e1d78ee59937666 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e005df5f2a6a73d6277a224443f6a1ae99f46dc4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:56:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b72753eb35420ee32878ba06982cbc768d1be5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:59:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32553bd54412423fab2d168a4182e350 : SUCCESS in 1m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e1e4ac52e7044ba7b039d488d6831717 : FAILURE in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/efdd8aa34c0541f78b171cf8ceebcf93 : FAILURE in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85d02112795999317c050b9b03aa3736e7ee84bd","revisions":{"85d02112795999317c050b9b03aa3736e7ee84bd":{"kind":"REWORK","_number":1,"created":"2021-09-30 05:42:49.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/76/811876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/811876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/811876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/811876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/811876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/811876/1"}}},"commit":{"parents":[{"commit":"6ec20f9929baa61445066cc9b79e8cb4d137f915","subject":"Merge \"Plumb allow_modify_pipeline through run_wsgi/run_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ec20f9929baa61445066cc9b79e8cb4d137f915"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-09-30 05:02:55.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-09-30 05:42:24.000000000","tz":-300},"subject":"Get rid of port to node assumptions and their modulo kludges","message":"Get rid of port to node assumptions and their modulo kludges\n\nWe had get_server_number for many years now, there\u0027s no reason\nto continue using the node_id\u003dport%100//10 thing.\n\nChange-Id: I5357a095110e8e4889c0468c154611209c6e8c07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d02112795999317c050b9b03aa3736e7ee84bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d02112795999317c050b9b03aa3736e7ee84bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
