)]}'
{"id":"openstack%2Fswift~812424","triplet_id":"openstack%2Fswift~master~Ie823091f93e76a4d658db7b1499f99d5faefcd81","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ie823091f93e76a4d658db7b1499f99d5faefcd81","subject":"cname_lookup: Work with dnspython 2.0+","status":"MERGED","created":"2021-10-05 04:50:39.000000000","updated":"2021-10-05 21:10:00.000000000","submitted":"2021-10-05 21:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812424","meta_rev_id":"355fea180b28a81a56611a1aed862b659159fcb2","_number":812424,"virtual_id_number":812424,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-05 21:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-05 20:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-05 20:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 05:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8cffb970aa864fcafdf292c4f20cf0420fb8a564","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-05 04:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8680a8e04d3126b21d649849879b89fcfdcc6e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 05:50:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624b752244b54fb68f319ce079821cc3 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/4ae5f05c4d734fd59ac2ae148550c9e3 : SUCCESS in 50m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/50f50eee4a8c45889f1836194030e36d : SUCCESS in 43m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b81cec0690094058a0cc59d552844840 : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8b030ae9daf14c0ebed50bb37c047716 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b89dcc005cd4418aad32be3501c51820 : SUCCESS in 15m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/041d6aa14d8d43eaacfb6f9322f61280 : SUCCESS in 13m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a08016deb6b14984af01641218a2c77d : SUCCESS in 16m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/40b6ba9db0644a01b6ac918f7ea27940 : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2e8e3af6a634dcca5cd7c0c9003184d : SUCCESS in 15m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2dd1ae5e411d48e2a4f63c9f13a21a34 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c187f9d22f44c38bc2b60190bd8d8bd : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ed508f1ed8248c084106883aba66cda : SUCCESS in 29m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3872151a15204afa8a4fe701c6658d44 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c568756f0edf4dab92121fa56d65212e : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/408604111bdd41998927c7228dc98591 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95ae5d67025948999d688d5cd59f0c80 : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6922738ea6e4bb38928b14b33be6fd4 : SUCCESS in 21m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dcdb3fa709784ac888ce927abfeb7338 : SUCCESS in 48m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7f79974c3f234d56b3403173a536313d : SUCCESS in 46m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/687fb64b75024fad9a98a1b6ddfa1bda : SUCCESS in 43m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f8b4da04ede4fdb8dd4ae5cbcfb17de : SUCCESS in 41m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6d83756d6c54b738b8678d68fd09945 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b05bdf6f4ec4e388597df838bedfdc4 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/667f251b16c14e90a9cc8e5dcc03e505 : SUCCESS in 29m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b58d67864444542a2df39ac42a806cb : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9409f30ee2a066693e5d47e01c1b7218ec185b7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 06:55:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9615de8c0bc4f8c9e9086d8ab52fee6 : SUCCESS in 43m 12s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0f32b7475d44362b1346ee80ebd70f7 : SUCCESS in 42m 06s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/46ec9cac667247aaa97d0d1d14aa21bb : TIMED_OUT in 2h 01m 34s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/950613602c3c4697a8528f1f8d803ec5 : SUCCESS in 44m 44s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f629c6055c94247b2f535876b2379f1 : SUCCESS in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c86b5cdbae80d191c4583fe929e4750713227737","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-05 20:08:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d30dffc7fdc9df883ed3e056a443f04cc67d684e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 20:09:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8829f7988bd0c79c7b7333fae4b2248e6f034bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 21:07:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2451f2f0e4fa4ee888737c564a659835 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/42d5586be7034243af93e789c4bb9486 : SUCCESS in 54m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da2f821619884950a61a90ef379c86f7 : SUCCESS in 37m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bee7ebdfd83a49beb788981e08584edc : SUCCESS in 9m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ca58e36430ae4c4aad9404efd6d75cd0 : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f2cfb127c414a3bb1627cd8aa4a2ee9 : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/864589a14f104db681af81e7341db8d0 : SUCCESS in 21m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c355a6528cf0412ea5cdeefe8b637e80 : SUCCESS in 21m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6bff0533ba8e4946b21a6cecde644b64 : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ccfda0db4d6946b49fe7b30a257eb910 : SUCCESS in 20m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/80af305fd7d84e40be67d88c0b31db29 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c735cabf4af74fe8bbc80c07d29765d8 : SUCCESS in 21m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/188a67ec5d424bc592537ecabc687eb4 : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/09b1d6d5ebba442c924a5b14bba4f573 : SUCCESS in 21m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/09992df1ab8f48b585cd39eab046cc87 : SUCCESS in 19m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/54d4326abc364843a8634aa57b263398 : SUCCESS in 21m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/624c4b3459c74f4da08c50273e3c1b5e : SUCCESS in 6m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39c5fe3e68c249108b51f45d9b7aab47 : SUCCESS in 53m 09s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d6f011aed4534c43886616bf9dcf774a : SUCCESS in 55m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/354c1125136d43a48899745b36e7c5a8 : SUCCESS in 40m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f4b5ddf4a6848d381a0de85b0b19816 : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7695560940b04865905394c795ca102d : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ab99d0ec8f4ff5967cd57e583d2a1e : SUCCESS in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97fa9e71a6064ae0ba462ae532124c29 : SUCCESS in 29m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc95a94865264880a0ca79070eecad29 : SUCCESS in 46m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9eeb11526bf31a33b5bd57e066e8ef0a2923ae3f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 21:07:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"355fea180b28a81a56611a1aed862b659159fcb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 21:10:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3718e79c49483ab538d3bd5487bc51 : SUCCESS in 1m 23s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9c7b2b3d78df4dc2b4e28e4d8e2e9274 : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9f10a3010273454fb4d7980117d89115 : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cde3938a9383b062630731b297d0abd9e054efa","revisions":{"1cde3938a9383b062630731b297d0abd9e054efa":{"kind":"REWORK","_number":1,"created":"2021-10-05 04:50:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/812424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/812424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/812424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/812424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/812424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/812424/1"}}},"commit":{"parents":[{"commit":"339cb10cfa1e04ba72dee9e03242c24b95edcc49","subject":"Merge \"Bug: fix s3api multipart parts listings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/339cb10cfa1e04ba72dee9e03242c24b95edcc49"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-05 04:50:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-05 04:50:26.000000000","tz":-420},"subject":"cname_lookup: Work with dnspython 2.0+","message":"cname_lookup: Work with dnspython 2.0+\n\n...provided eventlet is new enough to be compatible.\n\nChange-Id: Ie823091f93e76a4d658db7b1499f99d5faefcd81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cde3938a9383b062630731b297d0abd9e054efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cde3938a9383b062630731b297d0abd9e054efa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
