)]}'
{"id":"openstack%2Fswift~813715","triplet_id":"openstack%2Fswift~master~I923033e863b2faf3826a0f5ba84307addc34f986","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"17363":{"account":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"last_update":"2022-02-04 08:15:49.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-02-04 08:15:49.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-02-04 05:34:41.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I923033e863b2faf3826a0f5ba84307addc34f986","subject":"Fix multipart upload listings","status":"MERGED","created":"2021-10-12 23:37:13.000000000","updated":"2022-02-04 09:16:27.000000000","submitted":"2022-02-04 08:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"813715","meta_rev_id":"c0a8d58aabfcf29d2dd79ac7d50f4eaff6d27fc5","_number":813715,"virtual_id_number":813715,"owner":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-04 08:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-04 07:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-04 07:16:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 00:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-13 14:44:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-10-15 15:40:12.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-11-18 19:33:50.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-01-25 21:04:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"f57df8815314b35a486d587fc0698c82385fa4f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-10-12 23:37:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"360e8664fffe2180427a0ecc59fd3b3d2591d3af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 00:37:03.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8eacd63a5e147338e5501e4b65bbd68 : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f0f063ab59304071afbf3a7af73a0309 : SUCCESS in 55m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24962e737521449db8354d86e0ac0d79 : SUCCESS in 44m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8011e1c19f5b45819f1d132cc422f0f9 : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f99942eac8da4725bc5a9c022b85c06e : SUCCESS in 9m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c77c04d119f5449bb8d3cc01393f90eb : SUCCESS in 15m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a357a89f20548d9b7ab1fd34e07f44e : SUCCESS in 16m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1d118486dfe040f8a77f2ef379af9d6b : SUCCESS in 16m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/32650459215a42d79dc8aab37fcced53 : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3185a21700a9457daf515d2d8c262198 : SUCCESS in 15m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/65d0ee7d2f7c43c496e9f34b3c5f9484 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/629ecd4b65ef42929260681adf66af73 : SUCCESS in 16m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c051292d1eb6455e9c4bab0d241da6b2 : SUCCESS in 17m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2748a170f7e840c196907b7bf7bb65bf : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/065dba3f494f48d780236fc41069b2ff : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/134781e1965c4f80997022fbc34f8b1b : SUCCESS in 16m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68b76341384145bcb5ca17c22b298f56 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5cdc06c102f3493db4afd3019ef026fc : SUCCESS in 24m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9b4252a21def4368ac39cce5beb812cb : SUCCESS in 48m 32s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fb6ebebc096a4c4c9b7f293901c32171 : SUCCESS in 54m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa5ea0e9cf7546799902a3dc496fea7f : SUCCESS in 38m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0771639467174efcaf49b02f9ce7b48e : SUCCESS in 38m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/241a4207261d438d980236de5f1a8a74 : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7d70eeb2354f85a68550356c1d572a : FAILURE in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/251b8091d76c4b3997a9cbb5654d9838 : SUCCESS in 22m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76fa54f564f2458192886aa6200e62e8 : SUCCESS in 47m 35s","accounts_in_message":[],"_revision_number":1},{"id":"52470f3d3a33618a3366a654057440e718737053","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 01:42:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e9f89dc07fd43c4aaf52751caf748c4 : SUCCESS in 45m 14s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a740399d96ea4b13a2da245bcf96a178 : SUCCESS in 50m 50s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/e08490be4c484a4e975f0e9a3c9012da : TIMED_OUT in 2h 01m 30s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0feebad9ec94c0185e9bec019f83429 : SUCCESS in 48m 14s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/953a5095dc464e11b3323f0979447599 : SUCCESS in 47m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8e06783be625b326334a9a9e790e83abc4c43c8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-13 14:44:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"720dae25f1e4226b6a56a23530ade81a4484f597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-10-13 17:55:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16e2463c48aadbf39dd72be18082a6a0b312546b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:01:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8424452fea2d47e2a85fd1f2004dd7e8 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/a668189d2b2746f9b7509fde3b0236b2 : SUCCESS in 59m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d37fb42f6b8b4bcf97ad8a3e6c375a8b : SUCCESS in 38m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/53359cd817e942cca13521b494d7611d : SUCCESS in 7m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/096b85808c38468d91d8a976e5c78489 : SUCCESS in 8m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3bdbaf93c66b45ad98e5325de6882f70 : SUCCESS in 14m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d57df58714c84f2c909781ead7feb41e : SUCCESS in 18m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4e20c26f14fa4e1686f6080de8654b00 : SUCCESS in 19m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a45a40f2548e4e73a6347d44d9ed1eb6 : SUCCESS in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc1b92d576a2440cbaee41b0a1503793 : SUCCESS in 17m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f3ab722d6e284929b241fd29624cb23b : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a824ac88b2d4172956a5ba706d503fb : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b24780f5d9b64726898f82438d5e6893 : SUCCESS in 27m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/903f7ca688d3459398fea4b78832f7b1 : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9a598beed92e42deb2692a987ed6e7f8 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5b7741706428451ba60ca54c04fb29cf : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b025d6abed942a984aa801a74590e6d : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fb62142624234db8ab9b423946fcf36e : SUCCESS in 26m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/74e6b4892ed0476bb3c41c565d39cc53 : SUCCESS in 46m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/df073eea9f8f4f4dacec845ed7f32b53 : SUCCESS in 1h 02m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45dead83b6434dd1a0bc15f7b9c52ec8 : SUCCESS in 28m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f26bf70a50f4bae827c66ba16fd4cf3 : SUCCESS in 42m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a8ac1bd95844f988fabb9e135840e9e : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74c614e6b9f457b9b1ef871596bcbcf : SUCCESS in 7m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c3d636d3ae28418bb3d572de6cc86fe0 : SUCCESS in 21m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bf6acf2f5094671858b2b3a34526d5f : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":2},{"id":"e0c9d1cce677683c12f7228b2e95b38d9dcb9ea1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:52:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03c5c72317604a4da2dac73bac1749f3 : SUCCESS in 46m 51s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f48f34a4872e46b8827004a25267c088 : SUCCESS in 41m 23s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c150f059a0d74abb8e4a380e9af91961 : FAILURE in 1h 54m 18s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e878c2b0e29f4ee5a38a3895c55bf38d : SUCCESS in 43m 16s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f81918e0cf74750839352d64a061d44 : SUCCESS in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa41eb24794b16b4bb7a335bce2a4f830c34fe7a","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-10-14 16:07:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c01c75e093ab8f385260cf48deb88e39a7cae497","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-15 15:40:12.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9752ce14eade13855bb0f82094448d781f265382","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-11-17 00:18:26.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"319359e19cd6517c4d640f9f22e11ca48c37df64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-11-17 00:21:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ca6dde8c6380ca6300ef3d0ecb6f7fd24ba3ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 01:38:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b8fe431a9446c6b58e6c46439948da : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/16648d878c2a418d8797ce6031daf719 : SUCCESS in 1h 08m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec0a89dcd55e44f3a0f294e655eb55f7 : SUCCESS in 53m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f534f3d12f2437db7d708b39eca9971 : SUCCESS in 7m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0212d8b037234207b216aaacf346aa69 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae9b521e10214c4fa4655d6baffa0f98 : SUCCESS in 15m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/88eaa07864e043acbfde7fb01a5c2a52 : SUCCESS in 18m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b61aa03d2d104b608ca6c497673ebc7d : SUCCESS in 15m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3143fbf9e8b247fcba55979501d968b1 : SUCCESS in 16m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4fecfeeecff4508a535d16860109a02 : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a76dfe5390384419ae5e43b6108db08d : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d9a20fbbf1aa430abe522be87e3f8fa7 : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/352ff1a008e54040ac08ba5bf805ff7a : SUCCESS in 17m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b9d4630f83254469aebb649513089b0b : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0798be5bfc6846a7a1c05a0f68d077ec : SUCCESS in 18m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/afd37938617e476aac6e7ccf65de289f : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/97e27e8674704e74b7eeeba725655687 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1b17629631fa4412bb465ca15123dcbc : SUCCESS in 20m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c15232c95a634989a739fcff82b74a48 : SUCCESS in 58m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3b3493a5e092426e887fd3f08c5e8274 : SUCCESS in 53m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/679641b2c29845368e3202f366a30724 : SUCCESS in 39m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7ba80a1b3c14081b62633ba0101a85a : SUCCESS in 41m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58c7bf736aa1487d80b5e7ddac139231 : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/610b91329506491fae0b6cdeec8c3c85 : SUCCESS in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0119a3ba5f9a449c9155bfd463fd8a3a : SUCCESS in 19m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f019c2b139a4903b5949e8462012712 : SUCCESS in 1h 04m 46s","accounts_in_message":[],"_revision_number":3},{"id":"593d3c66c200310499e28a758db3bd820eb1f66f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 01:44:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdf6065a54ec417a9eebffcc9ac4a796 : SUCCESS in 41m 40s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b7e540a30e04aa188c10dc51d37bee4 : SUCCESS in 47m 23s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/09f5aaf1007e42138cf8b976655c6d69 : SUCCESS in 1h 17m 37s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/974edebda3d34076b4ecb9cc3ac2262b : FAILURE in 54m 02s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1996907a62ae409388277f353f76d3a4 : FAILURE in 56m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f2ce0e4755c90c1e7ead5c44cb729444422e8765","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-18 19:33:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"974b26ec51a11fb10633b8a36c9e70f7a70aa65d","author":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"date":"2021-11-18 19:44:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"62ca863f9c6cc6c057bca521c0e41af4ecf62d85","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-25 21:04:31.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ddeb8f4616dc89245277c6a48154b3f9b8d98182","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-03 00:34:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b54e80006d3a7438916fc06b6429fff6e278bee0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 01:22:24.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6a29d24f2334108841127ce9f42fb60 : SUCCESS in 30m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a08f7e3c42b544d4b1e3641cfe156aeb : SUCCESS in 37m 27s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/e2f2dd0567cd496aa91cb55948d029de : RETRY_LIMIT in 4m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c513cb12b2b4609b607d8b27641c292 : SUCCESS in 45m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/924abde4a8b548919a348aaaca0c69c5 : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":4},{"id":"f2d64b4ae10951b9fedce4d72230da7262967d0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-03 02:12:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ae310e3a39436ab524c87f4312c9da : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/5a774f6af07345c9acf51b1e021bacd3 : SUCCESS in 1h 01m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2173e3805171413098bd1e9afaf0d723 : SUCCESS in 54m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1d032d67f0a439583a8b9c4e5528ff4 : SUCCESS in 46m 11s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/686eed72d7734c6aba776b4d999cf82a : RETRY_LIMIT in 4m 49s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b9a6c9bb6d9c4e94a64a453176df2b5d : RETRY_LIMIT in 5m 53s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/08642d74c5dd4ea2be20fb471a887921 : RETRY_LIMIT in 5m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56f9e36543664f14b12b1a9e6e5c598a : SUCCESS in 8m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b2797708162b4e17a9fd1f5b1375c8a5 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d1c927b8102452ca652b816301e0195 : SUCCESS in 21m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76f9298e95fe4e0691a2f699ca006ab5 : SUCCESS in 19m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/386940903e1649ca899b5b3ed5ca43b3 : TIMED_OUT in 1h 31m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab52bfd3528f4b368629fe5ad2c23b1e : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38c271e6d8db4c16a4b3d41a4bad943a : SUCCESS in 16m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3536b03dc6dc42aa9cbfc1c83ea125fe : SUCCESS in 17m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8aef2fdebb11480da705e8321cb2dddc : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e8ac5acb344842f5b0ee4c4562f84481 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/79b088091a7e4415bd3a4649589cc9a2 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e821a357be8b476ab01133eec1d91ec2 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd783adc4c4a4673b582d2bbdd986e89 : SUCCESS in 23m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ccfe87fce9a468ca326d4c46be1a937 : SUCCESS in 46m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/71dbb1fd759d48cf86e09fc0b0149f44 : RETRY_LIMIT in 3m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/becb8cd24daa4137940207fdf980b5b9 : SUCCESS in 47m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db7013cf82eb4dfc840aca196dae1529 : SUCCESS in 38m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2995460878f848e599c5574ed1f3b724 : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84b41d283a844b79c9734a661d59ea9 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8e525fb8519f4d6fae8e3c0fe5f4904b : SUCCESS in 19m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b00176e553f341a5bcc067a82b6a6978 : SUCCESS in 50m 49s","accounts_in_message":[],"_revision_number":4},{"id":"bc5192b3448dd3f3b35beb12723d6415a3b2da44","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 05:34:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"243d96566e61ae856903c9eb6c52c62522818cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 07:05:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb8085d718c4094adb2a116b7bc36a3 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/de1e0a8c3a2f42428d0e4decc8ef778f : SUCCESS in 44m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7cb11b4f959b40cf80417b781e8e1a45 : SUCCESS in 40m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8f07c524f394c10abecb6a28ccdfd91 : SUCCESS in 42m 57s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7ba9ac46b39d4b8ba93fa5f7034a2aa7 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ce1c100ae350490b9fb7e0247fe85f65 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/caf98e83ebab49d39a388e5c4460d639 : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2060a424a811458fa993230a6ef5de04 : SUCCESS in 7m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/165d8793bbf94258bee9a35bc25951f3 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/43343375716644d4bf122996da7533b3 : SUCCESS in 14m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2152900c647c4d3ebe70108974ce0af7 : SUCCESS in 13m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/138c19430939414eaaa485bf5aaa05c2 : SUCCESS in 12m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2b445376beeb4acd9ca8bf8c7676d8b8 : SUCCESS in 22m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4c15a4de3044a488df9aa182afc043c : SUCCESS in 22m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/29b0be8b2ec648118a5e989bccdd1ec1 : SUCCESS in 15m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f97e24c3c15a47edb3a445b027b04932 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/28657e57e52e4d1ca07e82abeb3d1be1 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f7adba57ad14112bc41bb03899c9d21 : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/518c278f3b1d4d79b7054298275a8030 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e78c0f1ffe864779ab8c8574986c18f9 : SUCCESS in 22m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/64b97c36b40a426d978e48ecb9263bd4 : SUCCESS in 57m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/578fd7e0a5674bd9b7316364c4a7719f : SUCCESS in 54m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4920c4eb43304ace9c87e2d181f117d6 : SUCCESS in 28m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/310e84a5ba694204aefc2f28520da986 : SUCCESS in 33m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ece3c0ef06914f2fa7debc641f3ca843 : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f89206f7e29428399b44f1c070960be : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4dbc02007cff4b7482620af2666a1aa7 : SUCCESS in 28m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be7d84da47464fae8ef4b1b1956866e5 : SUCCESS in 51m 40s","accounts_in_message":[],"_revision_number":4},{"id":"5d9e0eacb3fc3c1fc74bd1ad3dc680ffd981480e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-04 07:16:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"db6e64a8b12cb0be276b88c6223b414b2ecaef33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 07:18:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"11cbb000c581559efb91d9141a92f857724f6b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:15:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36936b1027e4e2f8a5f066d306add94 : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/3edebc0b202848ee9962c555cc8b5e9a : SUCCESS in 51m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b696f74d4cc64c15ad8236573d39aef7 : SUCCESS in 33m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/265aff162d16468b9bad333fbd6d9fbe : SUCCESS in 37m 44s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5f5ad92f9b8947b4850a1eb5635551ef : SUCCESS in 8m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f5152d7901294156ba838fa7f6b69202 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/724d1350ce7f485389c0a86a903e25bc : SUCCESS in 12m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73f25e4dddca492abcbcd01ee272bcec : SUCCESS in 17m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97cc48bb6d8445b49b67c53d4ad9da58 : SUCCESS in 20m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cca82de46b174feeb8f8db00214975b8 : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a60bbe4d120e4fdb8173f0e9f792905e : SUCCESS in 24m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/53085ea6dc0c4b81a58d32d810c610b5 : SUCCESS in 16m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2a6aa388f88247a3b324fd9b943cebcb : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8cbe9e1afe2641e48c5c79d46d42c7ca : SUCCESS in 18m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/71b49b0f6f4d43a89679e59c167b51df : SUCCESS in 20m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d31c3640999946bc8a45ef2c717f2185 : SUCCESS in 5m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6dbf987a60a543c2a18750dbfaeee96f : SUCCESS in 51m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c835397a1cc429c8164f142e84e3fbd : SUCCESS in 52m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/41896bb1564e4be49fe83184bca33507 : SUCCESS in 40m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7958263afd0f477f8983165bd14f41fb : SUCCESS in 30m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edbb0329cc0a4c2e82664a8a1810adfe : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227e09a621414705be8a425e0627fc73 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a59dff13e2ba432c8c48e44d2b28e240 : SUCCESS in 20m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e2185e216434ede97e87950646988c0 : SUCCESS in 39m 38s","accounts_in_message":[],"_revision_number":4},{"id":"b471647fbbd404643871afcdf9a1e143445f9802","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:15:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"63e27c0f04fa7e20b9d19cce60ead7878c4329ad","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 08:18:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0d96953c0845399db6a23d01d4030c : SUCCESS in 1m 26s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0c8818e830c540c083e39c28dd5dc8e0 : SUCCESS in 1m 00s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9feebb961fbb47e291725b6b3da98ee1 : SUCCESS in 1m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c0a8d58aabfcf29d2dd79ac7d50f4eaff6d27fc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 09:16:27.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fc4b5953c6e48ea96e91f46fa5ebe99 : SUCCESS in 42m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23dd629b7dd24ae3b51713d4b31c63f8 : TIMED_OUT in 1h 31m 13s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/79de3530d4914dcd8da3eb9694097cc6 : TIMED_OUT in 3h 01m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c1dccf486e14384a38774790ce361e8 : SUCCESS in 1h 11m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f84da9d538743cfa6e725012afeefdc : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"363aa3319eda33af9699916cbc75cf5d779a7352","revisions":{"fd3c3a8eeb1f7d67384d9db8ab6bb0b5c6fcb1f5":{"kind":"REWORK","_number":1,"created":"2021-10-12 23:37:13.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/15/813715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/813715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/813715/1"}}},"commit":{"parents":[{"commit":"a5fbe6ca41fbc66a0952f9430744fc3f552b335d","subject":"ec: Use replication network to get frags for reconstruction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5fbe6ca41fbc66a0952f9430744fc3f552b335d"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-27 23:46:44.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-10-12 23:37:01.000000000","tz":-420},"subject":"Fix multipart upload listings","message":"Fix multipart upload listings\n\nWhen listing current multipart uploads, we have to keep listing the\ncontainer until we either list all the entries or there are enough MPUs\nto return to the caller. Otherwise, it is impossible to list all of the\nmultipart uploads when some of them have \u003e 1000 parts.\n\nChange-Id: I923033e863b2faf3826a0f5ba84307addc34f986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd3c3a8eeb1f7d67384d9db8ab6bb0b5c6fcb1f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd3c3a8eeb1f7d67384d9db8ab6bb0b5c6fcb1f5"}]},"branch":"refs/heads/master"},"5a7cf2aaa379f6d43eea468ef36b005020ed979c":{"kind":"REWORK","_number":2,"created":"2021-10-13 17:55:37.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/15/813715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/813715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/813715/2"}}},"commit":{"parents":[{"commit":"a5fbe6ca41fbc66a0952f9430744fc3f552b335d","subject":"ec: Use replication network to get frags for reconstruction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5fbe6ca41fbc66a0952f9430744fc3f552b335d"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-27 23:46:44.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-10-13 17:55:01.000000000","tz":-420},"subject":"Fix multipart upload listings","message":"Fix multipart upload listings\n\nWhen listing current multipart uploads, we have to keep listing the\ncontainer until we either list all the entries or there are enough MPUs\nto return to the caller. Otherwise, it is impossible to list all of the\nmultipart uploads when some of them have \u003e 1000 parts.\n\nChange-Id: I923033e863b2faf3826a0f5ba84307addc34f986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a7cf2aaa379f6d43eea468ef36b005020ed979c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a7cf2aaa379f6d43eea468ef36b005020ed979c"}]},"branch":"refs/heads/master"},"bdf16423d1c567af55dcd937a0c45db23d5a7e62":{"kind":"REWORK","_number":3,"created":"2021-11-17 00:21:17.000000000","uploader":{"_account_id":17363,"name":"Timur Alperovich","email":"timur@timuralp.com","username":"timuralp"},"ref":"refs/changes/15/813715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/813715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/813715/3"}}},"commit":{"parents":[{"commit":"a5fbe6ca41fbc66a0952f9430744fc3f552b335d","subject":"ec: Use replication network to get frags for reconstruction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5fbe6ca41fbc66a0952f9430744fc3f552b335d"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-27 23:46:44.000000000","tz":-420},"committer":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-11-17 00:20:55.000000000","tz":-480},"subject":"Fix multipart upload listings","message":"Fix multipart upload listings\n\nWhen listing current multipart uploads, we have to keep listing the\ncontainer until we either list all the entries or there are enough MPUs\nto return to the caller. Otherwise, it is impossible to list all of the\nmultipart uploads when some of them have \u003e 1000 parts.\n\nChange-Id: I923033e863b2faf3826a0f5ba84307addc34f986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdf16423d1c567af55dcd937a0c45db23d5a7e62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdf16423d1c567af55dcd937a0c45db23d5a7e62"}]},"branch":"refs/heads/master"},"363aa3319eda33af9699916cbc75cf5d779a7352":{"kind":"REWORK","_number":4,"created":"2022-02-03 00:34:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/813715/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/813715/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/813715/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/813715/4"}}},"commit":{"parents":[{"commit":"6e0ec0ad9cb5f561246c92c6f4b83a39e05dfd4c","subject":"Merge \"s3api: Fix non-ascii MPUs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e0ec0ad9cb5f561246c92c6f4b83a39e05dfd4c"}]}],"author":{"name":"Timur Alperovich","email":"timuralp@swiftstack.com","date":"2021-09-27 23:46:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-02-03 00:33:20.000000000","tz":-480},"subject":"Fix multipart upload listings","message":"Fix multipart upload listings\n\nWhen listing current multipart uploads, we have to keep listing the\ncontainer until we either list all the entries or there are enough MPUs\nto return to the caller. Otherwise, it is impossible to list all of the\nmultipart uploads when some of them have \u003e 1000 parts.\n\nChange-Id: I923033e863b2faf3826a0f5ba84307addc34f986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/363aa3319eda33af9699916cbc75cf5d779a7352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/363aa3319eda33af9699916cbc75cf5d779a7352"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
