)]}'
{"id":"openstack%2Fswift~813731","triplet_id":"openstack%2Fswift~master~I8c77c3c7e724768bf40d6d33bfbe72604fd4f6bd","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-10-13 18:43:33.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I8c77c3c7e724768bf40d6d33bfbe72604fd4f6bd","subject":"Bump up timeout on swift-probetests-centos-8-arm64 job","status":"MERGED","created":"2021-10-13 02:29:05.000000000","updated":"2021-10-13 19:45:14.000000000","submitted":"2021-10-13 19:40:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813731","meta_rev_id":"127dfb83364e2c363c31f3183da56f355430d4f2","_number":813731,"virtual_id_number":813731,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-13 19:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-13 18:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-13 18:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-13 02:29:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c613c6ff8e0fba4d009078643232744580f1800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-13 02:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"847ae6222fc301fd4c1da1f94b9de8e5b556ddf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 02:29:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  The job \"swift-probetests-centos-8-arm64\" exceeds tenant max-job-\n  timeout 10800.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: swift-probetests-centos-8-arm64\n      parent: swift-probetests-centos-8\n      nodeset: centos-8-arm64\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n        under Python 3 on top of arm64 architecture.\n      timeout: 14400\n\n  in \"openstack/swift/.zuul.yaml@master\", line 354, column 3","accounts_in_message":[],"_revision_number":1},{"id":"317a93f1a1a048d4df1fcbd26c576dd296b31aed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 02:29:48.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch master.  The error was:\n\n  The job \"swift-probetests-centos-8-arm64\" exceeds tenant max-job-\n  timeout 10800.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: swift-probetests-centos-8-arm64\n      parent: swift-probetests-centos-8\n      nodeset: centos-8-arm64\n      description: |\n        Setup a SAIO dev environment and run Swift\u0027s probe tests\n        under Python 3 on top of arm64 architecture.\n      timeout: 14400\n\n  in \"openstack/swift/.zuul.yaml@master\", line 354, column 3","accounts_in_message":[],"_revision_number":1},{"id":"0b38154f715dff0813111a913471830d7f2a7e2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-13 16:04:17.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"895726e67fe807c2752535e84e7246b04fb01236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 17:09:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b56f1a3d8c6543db8054abeb66b141c3 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/9cfd82dd15af4ccf9dca933ae8bb198b : SUCCESS in 48m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9d62f816761489688865a0486d8237f : SUCCESS in 36m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/459567f6aa4c4245851b7f74aefd6a13 : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cff4883e6a6a4858b371871887e62bcb : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bdf74e2eda746e1a3c08296935bf294 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/176d48f9796644719401dcabe9d4150f : SUCCESS in 15m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1ab2cdf9a5134d218cc5cf1f8210b38a : SUCCESS in 13m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/40e13ba7890c418cb3b0341342cc9a22 : SUCCESS in 25m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d35f7a21baf46b19c4968d19460e697 : SUCCESS in 18m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35cd3fc3956e4b2cace7e6b557b5626c : SUCCESS in 24m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fdc47c2bc7ab4478b679eb019255dde3 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b52abe3a02ed4e57b0414b798b2a1f94 : SUCCESS in 21m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/13013499e9e944cb9a06dfaf4de95df7 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d18b57c9a16b4090a2fc3938aa49f58f : SUCCESS in 17m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3ec4b7817fd48e8994a7c5ecda2fcea : SUCCESS in 18m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/94b3849df70c44e88eee3642fd60c1fd : SUCCESS in 8m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eead3bfed987443699f7bf8d9d85ff59 : SUCCESS in 25m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eea552f4c8714fbd966603c172cc6573 : SUCCESS in 55m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5265ae25df9c46dd9db6fc6df5f9e900 : SUCCESS in 56m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/015a5fe6dc424590b311b31a25396749 : SUCCESS in 42m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e3dd0c4258504f1db81b70c5da61c4da : SUCCESS in 44m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b51baeb97116492a98f5ffc46a1d771c : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c9598e15974b90bf42e0b8d24cb254 : SUCCESS in 7m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b48289522714ee6980b4f9e69a85fd3 : SUCCESS in 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5c7577c84694a8cb979917383452ef7 : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":2},{"id":"eddfbc360a6e58c02aae118cbb28f67d51ead6e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 18:00:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e533a2c48b44705b523f996007bcb41 : SUCCESS in 43m 59s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b534f77cd5a48a8a9f090fd658a6d4c : SUCCESS in 41m 40s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/00cc0ae0fcd04a7ab15dae9724349787 : FAILURE in 1h 50m 06s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b31a903b72a42dbbd6e36d44090c989 : SUCCESS in 44m 45s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7417b8d39ca14cff960a718bf60db1ea : SUCCESS in 44m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6679e75723216348ecac439857e21ef95533e471","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-13 18:43:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b88d9710eb3d98d138d417c6da9c282c0f92797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 18:44:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a739a7ec6424a7672c465b48468b8472885ec893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:40:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4240b6390c418a9e474535247fca98 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/6fc1c6d34af644f38fdfe7f8d3fdb966 : SUCCESS in 44m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c7c31a5609d4bdb99f34a337e90386e : SUCCESS in 43m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/05cce1105e7f4a4aa08bc6baab1ce222 : SUCCESS in 10m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d4734335782d45d9ab48dc2d8e070f0b : SUCCESS in 10m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0b5e11630b404a49a797a505845e1daa : SUCCESS in 15m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/32b0fda7b5604ce280083212af1c9132 : SUCCESS in 15m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a9b17577d0eb469fb02b7f6f83b6fac8 : SUCCESS in 16m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d14a9285cfcb4c4eb0bbc8fa7d2bde59 : SUCCESS in 16m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d6234b45df84d79aeb652e5fb4e636d : SUCCESS in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e34bfa7a525047488c4ff30a067eae77 : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2eca997795b844f6bd10ffecee7fd1a6 : SUCCESS in 23m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2fef1c7efc6d4642b313f57729a47f5d : SUCCESS in 17m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a30e52daab3d435a95f30598f86d7633 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2a9528766e964f4c852382e04be2db1c : SUCCESS in 20m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0a2aebf9c5a24ee087d8bd0066d2d1fd : SUCCESS in 20m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d147abc1932c4d97b215eb077b162617 : SUCCESS in 6m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d5865d060f2244c8a90716a49c4ce777 : SUCCESS in 53m 02s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7658d47c677347938f5b3cb508a6a05a : SUCCESS in 52m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5095e7417eee45c08bb6afbece3cef42 : SUCCESS in 40m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9d71d48c361d438cafd352b1186cfb10 : SUCCESS in 41m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40997eb6112444d29a807f349ed275e7 : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9056807e42ef44bc8b4bd9a8dcdb85d0 : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77dabac001d840df895529d959cdf7cf : SUCCESS in 25m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27fc43a37fc64688b4e2e3259f6160e8 : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d89567bf4ff724c8415809ddd0b4cdc3b6a3749e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:40:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"127dfb83364e2c363c31f3183da56f355430d4f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 19:45:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b71e38e22f6d4014a081d3847c521705 : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a738a59f0baa4e59a1b90326b52373a9 : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f32c730607974fb49ee5dfc009fe7d64 : FAILURE in 50s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f84e77a62a67d09ea780ccee4a8d235c74944ef","revisions":{"1f53d602559443d64c2583cc9cadab4eda445cb0":{"kind":"REWORK","_number":1,"created":"2021-10-13 02:29:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/813731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/813731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/813731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/813731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/813731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/813731/1"}}},"commit":{"parents":[{"commit":"32cc19962485d2a46d9c5a5f46cfb82234c8b6d9","subject":"Merge \"sharder: Make stats interval configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32cc19962485d2a46d9c5a5f46cfb82234c8b6d9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-13 02:28:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-13 02:28:51.000000000","tz":-420},"subject":"Bump up timeout on swift-probetests-centos-8-arm64 job","message":"Bump up timeout on swift-probetests-centos-8-arm64 job\n\nWe have seen a fair number of timeouts on this lately.\n\nChange-Id: I8c77c3c7e724768bf40d6d33bfbe72604fd4f6bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f53d602559443d64c2583cc9cadab4eda445cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f53d602559443d64c2583cc9cadab4eda445cb0"}]},"branch":"refs/heads/master"},"4f84e77a62a67d09ea780ccee4a8d235c74944ef":{"kind":"REWORK","_number":2,"created":"2021-10-13 16:04:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/813731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/813731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/813731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/813731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/813731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/813731/2"}}},"commit":{"parents":[{"commit":"32cc19962485d2a46d9c5a5f46cfb82234c8b6d9","subject":"Merge \"sharder: Make stats interval configurable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32cc19962485d2a46d9c5a5f46cfb82234c8b6d9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-13 02:28:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-10-13 16:04:17.000000000","tz":0},"subject":"Bump up timeout on swift-probetests-centos-8-arm64 job","message":"Bump up timeout on swift-probetests-centos-8-arm64 job\n\nWe have seen a fair number of timeouts on this lately.\n\nChange-Id: I8c77c3c7e724768bf40d6d33bfbe72604fd4f6bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f84e77a62a67d09ea780ccee4a8d235c74944ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f84e77a62a67d09ea780ccee4a8d235c74944ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
