)]}'
{"id":"openstack%2Fswift~814189","triplet_id":"openstack%2Fswift~master~If1909bb7210622908f2ecc5e06d53cd48250572a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-08-24 21:33:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If1909bb7210622908f2ecc5e06d53cd48250572a","subject":"Add MPU to s3api tests","status":"MERGED","created":"2021-10-15 15:40:01.000000000","updated":"2022-08-24 21:34:37.000000000","submitted":"2022-08-24 21:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814189","meta_rev_id":"0e47ca5fec1772d8b14f41b5d2cd9441aba5a9dd","_number":814189,"virtual_id_number":814189,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 21:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-24 19:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-24 19:19:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 16:49:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 19:19:44.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"069c54610c89e56cd1ef23312e71cd6625288b71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-15 15:40:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfccda64d27047dde48cb35ce5563b1c7160af13","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-15 15:42:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"501a3e5f8a6c68ecfb5855d9958867bbe6341cd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 16:49:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5681a031ecd4f76b96628e233792e4c : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/c92d07fe48a94fbf83b999517b4d134f : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5df7cdf72e74d16891b2f46633b8def : SUCCESS in 49m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f4d5de1ea9e4d948e9977bae03cd0e5 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a807c1c2fb5b41b4a140997b7efa113c : SUCCESS in 10m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/782ecc65be9a4d7a8607e3009ab694c7 : SUCCESS in 26m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/45708e30188446a88bfe1e1fe8fcb870 : SUCCESS in 17m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/69ba4a7729c048119c687247322a6506 : SUCCESS in 16m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/40cdf2a0d3b5438f8f4ed72ec62ba8d9 : SUCCESS in 17m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1068faaca85645e6a9ab00528ce42570 : SUCCESS in 26m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/743154f7bca84831a71e83d2dd1cc02c : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0e9961b16874374967f6cd7ff3a4684 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7778611bf1694f8d9ae1d664ab2c2469 : SUCCESS in 24m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b428cba26de04e0cac2ee6ddcdc28934 : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec0b93bdb9a7480b83b40d4879e60dd1 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/72aae598a0044d90b0e3c86681137db3 : SUCCESS in 23m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0091ef7d847344feb27ebc02a22006d2 : SUCCESS in 8m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/840f4ff1b8cc42c2beaffd2aa0f74b16 : SUCCESS in 28m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d20128ec4314448683915f0c5033d20b : SUCCESS in 52m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7c6c2dbb823e4a888559e3f36e92bd87 : SUCCESS in 56m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ca9000a4633417e814f69a487a40fb6 : SUCCESS in 45m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3cc27d86bd8449deac0db6c93dd9a65e : SUCCESS in 43m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6b7b4482606406b9570bceca20df96c : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e66431dca145a8a2bd67ee31e22ed8 : FAILURE in 6m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f33f343fd50247149f934c737a6cc75f : SUCCESS in 32m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2789c8621fca40ce8767fa0637c27bde : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4c0e22a6b3edb2396a40b51b59b41bdacb05f8ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 17:51:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c382fd8606da44d5af1a1e10c32b298f : SUCCESS in 1h 13m 30s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2c5c4b0835544f08e4365d30413b09f : SUCCESS in 58m 02s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/b3f70ffc62f04da28754cfa63f92599d : FAILURE in 2h 06m 20s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b81ef1ba040e48eb8b017ed6a9290916 : SUCCESS in 40m 56s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b2a3fcab53b475fb6ac28a2dcefc3a9 : SUCCESS in 1h 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c42897b1217cec928cae58f2689cc0cc126bb1b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-24 18:05:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc91bd25aa9669f8495b62aaf1e87f4608ea217b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:52:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6084d36651bb42e084a5f49408d0de78\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c59e1dddb9044aeabeb8238f9fd23daf : SUCCESS in 28m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/492f20c518764ecdb7bba162a82ce6bc : SUCCESS in 29m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6170ed000d564ff9a9f5206a4f574aa6 : SUCCESS in 45m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ed344de527d42fe8e9944f23a578eb5 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":2},{"id":"be58a96950d9749ec4b39339e89fe74e1595311b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-24 19:18:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7f0144bc2e839f6da2e3fd5bef6e4886609fc0a7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-24 19:19:44.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6fc32a9cb357cd37905a16bd0fc2cf9e37e3cb7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 20:04:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5ede3b5b7844091918c7f6fa9fff5b1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d945a89d00a4cdc81b4713799e2ae69 : SUCCESS in 29m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9469adf1133f4708a721cd7303fcfbf6 : SUCCESS in 44m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fd69b0908d642ab8436fc6427c51558 : SUCCESS in 29m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa3e93d52d6d4ec59c99955aa1d7a07b : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":3},{"id":"8384e1c8a6d3fdd72262c2b60864c8ab336ed988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 20:19:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2220994750834a8e89ee280f6c076bab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04dcf60e0ac45a6ac387dd04bb6401f : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/d5202e1742414f4aa30418fec9058712 : SUCCESS in 48m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/759089bedc4d4167af127b49afbefcb4 : SUCCESS in 52m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85b04f6160b24d6bb5dd396dee5ed459 : SUCCESS in 50m 55s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d5421217341b43b09d128b035d5464c3 : SUCCESS in 24m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ce975e73ef6a42a5abe90f6c9ed6edbd : SUCCESS in 23m 59s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eed0b7c42018467d965ed245cf436878 : SUCCESS in 24m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a61151a3dc6f449f9258bad31a2f7347 : SUCCESS in 10m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e06be9eda2cf42b1999038cc6d5b8c3f : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0fbfb12a902a49f5888af31d22abab98 : SUCCESS in 21m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c798c92aac14feaa6c6da44f3bc9e9b : SUCCESS in 16m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0f9b88c492484af4851dc3bce1449b2f : SUCCESS in 12m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b6a017f8518403cb95500ac3249a3b4 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c7be9e81ad4d400fab266427b9727800 : SUCCESS in 17m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aef933f7c2524ce7882c9e19db4de4cf : SUCCESS in 20m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/954983a784c24ecb8eb76448af073e23 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/10d72cfdf52d4836bcd715327a40acb3 : SUCCESS in 12m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a9bd1a0654c848ecb2e9a3ba70ef0b54 : SUCCESS in 14m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f8529678a604ad5a69839f35029685d : SUCCESS in 9m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/596b34fc5e2f46b68e537437da6505bd : SUCCESS in 15m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d385c1f8b08c45cc8776fcd2e7ce083d : SUCCESS in 9m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/00d5c2560d434f1b861b1b5411228864 : SUCCESS in 45m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/44d7c19807cd41c1806184be3f8be519 : SUCCESS in 47m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a993fa08f5342ae8714e6523afad19f : SUCCESS in 32m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad25a09f009148d6983aa4d81066a8d9 : SUCCESS in 32m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a624048eda34175afec7932f90c541e : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca21012185940bcb9dd7005ec9345db : SUCCESS in 14m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e21167c64d3462d9ea5aabf1d5935bb : SUCCESS in 24m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db316fe352ca403d94c3126ab73136eb : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":3},{"id":"84e8f986138e23a71ffa9b731217906fc684adfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 20:20:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"64610e9328f8e801a2b2683670b6348bd1c2aeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:33:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7e73dbd2204f32a5173703536347b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8044f6dc64745038e2e389e4263d785 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/22e5c562901342aa8899a997a4b025fd : SUCCESS in 53m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62c597b7428e4800ab2122a3ecef27f3 : SUCCESS in 45m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a79613b9d8947fbbc4ebbbf29ff17b3 : SUCCESS in 48m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5407363f308a44cf8da5a40c538e2f25 : SUCCESS in 8m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/421a8670003a444299073a7f05147c11 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/128c16108fd543d5b0c18d2d7fc00ec5 : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f35574887342490d88fea624ed2c1174 : SUCCESS in 18m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b354ff7752594fcfbcbe164dcaf521ed : SUCCESS in 16m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7a2273d57604831b257f56a8a1682f9 : SUCCESS in 19m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff9dd2b2b01f4d3da699ff66fcbe969c : SUCCESS in 22m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5572d8ccbf2483aa3502b78e6e1ff3c : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fddb683c53004fa982d6bd57d34d3699 : SUCCESS in 20m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/954a406610dc47e8825ed856984ce541 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2e01a50191814f03819f5c87678d16c1 : SUCCESS in 20m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7cfcae2e37e4fe08c33e3c6e2a14cd3 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7280633d95494d1090d81eb39ccfbba1 : SUCCESS in 8m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8ca43cfdec3c43a08a2ccec824b6b295 : SUCCESS in 1h 12m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8c5bb06916c4f14ab976c792aa4d478 : SUCCESS in 48m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d565030d446b4acc86eb17ad34a55897 : SUCCESS in 47m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9d383bb0771a472ea4e46a98a248444a : SUCCESS in 50m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fbb0c416b4549a18b562afe869c997f : SUCCESS in 22m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f5e86756d7944a8bc946670047dabad : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea8d460584de4a05ae7f1db103bdda19 : SUCCESS in 30m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/117a8805147740e7822a7f8acde847c1 : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":3},{"id":"bbc385b84d46a1e8cda9c3da8fa97629ae0126ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:33:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0e47ca5fec1772d8b14f41b5d2cd9441aba5a9dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:34:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f1cc829863490e94f41ff9176637d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d19fd7c6960f4cb3bbea9062e5d8cb5b : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/de84f6cc17f74c4396fc3ddd55bc4aa3 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a724a98129a845a9999e5101bc1848b2 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0feee1900efcca0b70b185e1bc34dec688b974d","revisions":{"5af5652c50e19b7e256ba1ade2994098792b83b7":{"kind":"REWORK","_number":1,"created":"2021-10-15 15:40:01.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/89/814189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/814189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/814189/1"}}},"commit":{"parents":[{"commit":"5a7cf2aaa379f6d43eea468ef36b005020ed979c","subject":"Fix multipart upload listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a7cf2aaa379f6d43eea468ef36b005020ed979c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 15:33:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 15:39:38.000000000","tz":-300},"subject":"Add MPU to s3api tests","message":"Add MPU to s3api tests\n\n... and reword some mpu listing logic\n\nRelated-Change-Id: I923033e863b2faf3826a0f5ba84307addc34f986\nChange-Id: If1909bb7210622908f2ecc5e06d53cd48250572a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5af5652c50e19b7e256ba1ade2994098792b83b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5af5652c50e19b7e256ba1ade2994098792b83b7"}]},"branch":"refs/heads/master"},"b3a9299093e3a1c59395bd81f428bac8a87b430a":{"kind":"REWORK","_number":2,"created":"2022-08-24 18:05:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/814189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/814189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/814189/2"}}},"commit":{"parents":[{"commit":"6504a5f59e84c3d56409836af522866c414142ca","subject":"Merge \"swift-manage-shard-ranges repair: ignore recent overlaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6504a5f59e84c3d56409836af522866c414142ca"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 15:33:34.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-24 18:05:27.000000000","tz":-420},"subject":"Add MPU to s3api tests","message":"Add MPU to s3api tests\n\n... and reword some mpu listing logic\n\nRelated-Change-Id: I923033e863b2faf3826a0f5ba84307addc34f986\nChange-Id: If1909bb7210622908f2ecc5e06d53cd48250572a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3a9299093e3a1c59395bd81f428bac8a87b430a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3a9299093e3a1c59395bd81f428bac8a87b430a"}]},"branch":"refs/heads/master"},"d0feee1900efcca0b70b185e1bc34dec688b974d":{"kind":"REWORK","_number":3,"created":"2022-08-24 19:18:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/814189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/814189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/814189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/814189/3"}}},"commit":{"parents":[{"commit":"6504a5f59e84c3d56409836af522866c414142ca","subject":"Merge \"swift-manage-shard-ranges repair: ignore recent overlaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6504a5f59e84c3d56409836af522866c414142ca"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 15:33:34.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-24 19:18:03.000000000","tz":-420},"subject":"Add MPU to s3api tests","message":"Add MPU to s3api tests\n\n... and reword some mpu listing logic\n\nRelated-Change-Id: I923033e863b2faf3826a0f5ba84307addc34f986\nChange-Id: If1909bb7210622908f2ecc5e06d53cd48250572a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0feee1900efcca0b70b185e1bc34dec688b974d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0feee1900efcca0b70b185e1bc34dec688b974d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
