)]}'
{"id":"openstack%2Fswift~814196","triplet_id":"openstack%2Fswift~master~I3c944ff99213e1e828cfefcffc0cd714517c8367","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2021-10-15 17:35:24.000000000","reason":"Tim Burke replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I3c944ff99213e1e828cfefcffc0cd714517c8367","subject":"Add ringbuilder CLI tests","status":"NEW","created":"2021-10-15 16:25:46.000000000","updated":"2021-10-15 17:37:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8260ab208bc9a6a49409cb8ab7f15670dc41f08","_number":814196,"virtual_id_number":814196,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-15 17:37:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-15 17:35:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-15 17:35:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 17:35:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-10-15 17:37:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c80e12dfca4e4c77c41f2608f682d699737a965","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-15 16:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ae87c06615df1979cc695393558ba2b3cf1b167","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-15 17:35:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8260ab208bc9a6a49409cb8ab7f15670dc41f08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 17:37:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd26972e30f44527b72f0ab80a3b9f56 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/43853750935b4060899602dc9022a293 : SUCCESS in 1h 04m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5caca978a554270b17a3e2e82836cc4 : SUCCESS in 33m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be225559dd564ce29eb0b0428183ba27 : SUCCESS in 9m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5776440cbdfb4fd9a833fa63ca7d5e27 : SUCCESS in 11m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24589d53307b41ff839fd76b40bf08f4 : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33c56bcaa7d54e4eb717a475056c72c9 : SUCCESS in 16m 59s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/684cfc424f004d72b76920a7c172eb7f : SUCCESS in 17m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/05f00a1dbf1844d089fde07d2eb9e6d7 : SUCCESS in 20m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/60f69b20199c42519c44abf74e9bb504 : SUCCESS in 17m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f9b4da20e9044772bf48472df16717c2 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1be104d902b3460fad582261c8e6cfac : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7a013350b8dc460eab1ab969585993c8 : SUCCESS in 20m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8a2a37d313e8455b87e870b66f8f3387 : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/632cabd3f7d84c3f939a883eee42693f : SUCCESS in 15m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9ff206073e3443ffaaf3e3d4cae89413 : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca60ebd0efda456db90465abd81ed766 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dcb88014f5694b04b2818a6327928b4e : SUCCESS in 19m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/592ca8c4f06e4d708439b01c3f2e05ef : SUCCESS in 50m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b0ec8f2d8c3d4141a5d1164566aa7365 : SUCCESS in 52m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08c776d80732483ba70770d9dd183e13 : SUCCESS in 40m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/345701aa3a714be58198b526e6a740f4 : SUCCESS in 39m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90d3b170084c4d90bdf91e0c383d7ca7 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9618ff573cbc4316982a4268bb6aaa78 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02495fe2837049969689a19d4241a417 : SUCCESS in 33m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9dce949d0d5e4a9ca267821d947dc9c6 : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbb4c3486beafdcf242cd0023a38498f847ddabf","revisions":{"dbb4c3486beafdcf242cd0023a38498f847ddabf":{"kind":"REWORK","_number":1,"created":"2021-10-15 16:25:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/96/814196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/814196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/814196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/814196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/814196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/814196/1"}}},"commit":{"parents":[{"commit":"73c17515db7a08cc800106a4b54535512effac19","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73c17515db7a08cc800106a4b54535512effac19"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 16:00:19.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 16:25:24.000000000","tz":-300},"subject":"Add ringbuilder CLI tests","message":"Add ringbuilder CLI tests\n\nRelated-Change-Id: I689717272b56522ef1d830e07d3154509dafa5ce\nChange-Id: I3c944ff99213e1e828cfefcffc0cd714517c8367\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbb4c3486beafdcf242cd0023a38498f847ddabf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbb4c3486beafdcf242cd0023a38498f847ddabf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
