)]}'
{"id":"openstack%2Fswift~814197","triplet_id":"openstack%2Fswift~master~I963a5f878a719ab2c56f5683262efb6f5addb39e","project":"openstack/swift","branch":"master","topic":"ring-v2","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2021-10-15 17:24:47.000000000","reason":"Tim Burke replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I963a5f878a719ab2c56f5683262efb6f5addb39e","subject":"Test increase part power with v2","status":"NEW","created":"2021-10-15 16:25:46.000000000","updated":"2021-10-15 18:13:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"71361e345b056f1b6aab7a5c706c732a49b985cf","_number":814197,"virtual_id_number":814197,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-10-15 17:36:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-15 17:24:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-15 17:24:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 17:24:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-10-15 17:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f50cddd352f39478754205d8d4b6f1aeb58c49ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-15 16:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d7678f8a33379df51f411dd6903af0fca2e5b90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-15 17:24:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b641337cdfda9b22b18a3ae5af4ca963a8d9f8db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 17:36:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27212cb9fcc1423f931eaa042dc6ecdf : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/e7591b9e736c4952a133fd133ad97c82 : SUCCESS in 58m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2c1f67aae26405ba026329e4067db8a : SUCCESS in 1h 02m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a6b226db26f409db8f3c0b935a8dea5 : SUCCESS in 8m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/da4b8c6d36ec4946a45cd6893f053142 : SUCCESS in 10m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5c09645be0c24eafb2bde1de3ad2934a : SUCCESS in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e99f10242de14c0e8acda34b56ac746a : SUCCESS in 17m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71afad65a92448ecbd0584a2abf4fea5 : SUCCESS in 16m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5a671834049a431991949f88895afcf7 : SUCCESS in 16m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fae7c6efec004537a810b2187bade9c8 : SUCCESS in 19m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e1f86fee7974dda9d3c14b3faca2e5a : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae0a49940d9b40588788fc9eb929cc90 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/541b4feec03b464cad324fc5c80bf394 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/575e4577b5534607b691a36f441b1429 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1571a7055e24026a493927ae1fdf599 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/02ab6ffb315b4b4c913a3bc588dbffea : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1047dd9038c440b3b58580dcf6029ad7 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/37a1719297e6474da66190f817d1c80d : SUCCESS in 17m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1cec6fda5c9d49e1a820638636d5896d : SUCCESS in 47m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/dd2146a75b314055a102eaa1cbe0fb6a : SUCCESS in 54m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9bc0f3ae00f419cbde59071e81df528 : SUCCESS in 33m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bedb24c285140ccb1ed7b9fb16358c2 : SUCCESS in 29m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c551500698ac478db62d1aa488e587be : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567fdd4a0f054e8c87174d6e46a04d63 : FAILURE in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b893dcbb25864995aa471063a10db665 : SUCCESS in 23m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/271574fd6b704fada68f9a27a3c31c74 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":1},{"id":"71361e345b056f1b6aab7a5c706c732a49b985cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 18:13:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd87024da028453c95fa74414a9b196c : SUCCESS in 41m 34s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/076b943185464116acf47aa5c69176f1 : SUCCESS in 41m 45s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/e477bb29f292421d8d85e9ef971e7530 : SUCCESS in 1h 33m 31s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a59ae55b95d4509ab2d7a90fa9aabe8 : SUCCESS in 42m 41s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bd88beb807a48cb80ff1b163a567222 : SUCCESS in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee38e0b08838d7ff6d624303f9aad0b98859d2ea","revisions":{"ee38e0b08838d7ff6d624303f9aad0b98859d2ea":{"kind":"REWORK","_number":1,"created":"2021-10-15 16:25:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/97/814197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/814197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/814197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/814197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/814197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/814197/1"}}},"commit":{"parents":[{"commit":"dbb4c3486beafdcf242cd0023a38498f847ddabf","subject":"Add ringbuilder CLI tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbb4c3486beafdcf242cd0023a38498f847ddabf"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 16:24:50.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-15 16:25:34.000000000","tz":-300},"subject":"Test increase part power with v2","message":"Test increase part power with v2\n\nRelated-Change-Id: I689717272b56522ef1d830e07d3154509dafa5ce\nChange-Id: I963a5f878a719ab2c56f5683262efb6f5addb39e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee38e0b08838d7ff6d624303f9aad0b98859d2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee38e0b08838d7ff6d624303f9aad0b98859d2ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
