)]}'
{"id":"openstack%2Fswift~814510","triplet_id":"openstack%2Fswift~master~Id2c842cffad58fb6fd5e1d12619c46ffcb38f8a5","project":"openstack/swift","branch":"master","topic":"update_sharding_doc","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2021-10-19 20:23:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2c842cffad58fb6fd5e1d12619c46ffcb38f8a5","subject":"sharidng: update doc to only mention auto_shard experimental","status":"MERGED","created":"2021-10-19 01:56:53.000000000","updated":"2021-10-19 20:25:31.000000000","submitted":"2021-10-19 20:23:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"814510-update_sharding_doc","meta_rev_id":"d64fa7aaf998852fb771da7f841a38fe6bccfe41","_number":814510,"virtual_id_number":814510,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-19 20:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-10-19 16:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-10-19 16:20:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-19 02:09:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-19 16:20:25.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"422a8bde46d4dfe2c502d67b014a264b37de7db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-10-19 01:56:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cabded185650a310b407e429654a13677ef4d968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 02:09:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a19c12fb5fe404499b59a37a6ebb837 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2351637d1f82c75581558b208132ef96e60af453","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 03:42:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/6acbe6ec27324d2f9a1ff8e5d8c4e411 : FAILURE in 1h 43m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b890daed5258695d072d593b22a1ffecbf8c6a8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-19 16:20:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d742965fc779f40e7ef8c7039a830851b9f430c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 16:24:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bb153e0809deb29bbe79137762ea10d7ca3092f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 17:23:00.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d6af4427e49166bb8bf26584e83c4f6ab66375c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 20:23:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76353a57f25f4d1d871c269dc712d76b : SUCCESS in 10m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f512e5d8ff4d42d099978b0fe27862fa : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":1},{"id":"203d3a32b1980e16309d3fc85c9b64c9b8d14168","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 20:23:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d64fa7aaf998852fb771da7f841a38fe6bccfe41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 20:25:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38e29e2c2799484b9f7e913adf3ca3d4 : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/91e6ff8a5ecc4fc3bf3c11b670b22d13 : FAILURE in 1m 09s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b1b27c93155d4ff08f30ecd44f6746a0 : FAILURE in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f5e34a405822773b8c1422aad4ff64b5f645d14","revisions":{"7f5e34a405822773b8c1422aad4ff64b5f645d14":{"kind":"REWORK","_number":1,"created":"2021-10-19 01:56:53.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/10/814510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/814510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/814510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/814510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/814510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/814510/1"}}},"commit":{"parents":[{"commit":"dbd0960aeebedc0487699d3ca2a4d6f21e7ed524","subject":"Merge \"Update master for stable/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dbd0960aeebedc0487699d3ca2a4d6f21e7ed524"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-10-19 01:53:04.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-10-19 01:53:04.000000000","tz":660},"subject":"sharidng: update doc to only mention auto_shard experimental","message":"sharidng: update doc to only mention auto_shard experimental\n\nThere are been members of the community running sharding in production\nand it\u0027s awesome. It\u0027s just the auto-sharding swift of that remains\nexperimental.\n\nThis patch removes the big sharding warning from the top of the\nsharding overview page and better emphasises that it\u0027s the audo_shard\noption that isn\u0027t ready for production use.\n\nChange-Id: Id2c842cffad58fb6fd5e1d12619c46ffcb38f8a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f5e34a405822773b8c1422aad4ff64b5f645d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f5e34a405822773b8c1422aad4ff64b5f645d14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
