)]}'
{"id":"openstack%2Fswift~814778","triplet_id":"openstack%2Fswift~master~I41ad14a08ab73f275cbde3abcf3938c0e5f3698f","project":"openstack/swift","branch":"master","topic":"docs/first_contrib_opendev","hashtags":[],"change_id":"I41ad14a08ab73f275cbde3abcf3938c0e5f3698f","subject":"Change github.com to opendev.org so git-review works","status":"ABANDONED","created":"2021-10-20 13:27:34.000000000","updated":"2021-10-21 05:54:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13b6df10703d49ad98919111347d08f2eb14e2b1","_number":814778,"virtual_id_number":814778,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-20 13:38:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 13:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad0302db2bd14badbc2eb5f64b5e3ed3359d5395","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-10-20 13:27:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd1093eeb4fa782bd4ced74b65604c024c3b38e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 13:38:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07eb9c0148d844f392cde20287eaaf6b : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":1},{"id":"642a656f2eb77f2178e94cdbbbcb7f758dce1f44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 15:31:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/692202e7da074af389a1fec61b8d2e0a : SUCCESS in 1h 55m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13b6df10703d49ad98919111347d08f2eb14e2b1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-10-21 05:54:11.000000000","message":"Abandoned\n\nThe bug seems invalid.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d622770b5a96989d47fc521e163b0670eb6584b","revisions":{"2d622770b5a96989d47fc521e163b0670eb6584b":{"kind":"REWORK","_number":1,"created":"2021-10-20 13:27:34.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/78/814778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/814778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/814778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/814778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/814778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/814778/1"}}},"commit":{"parents":[{"commit":"7f5e34a405822773b8c1422aad4ff64b5f645d14","subject":"sharidng: update doc to only mention auto_shard experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f5e34a405822773b8c1422aad4ff64b5f645d14"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-10-20 13:23:20.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-10-20 13:23:20.000000000","tz":-300},"subject":"Change github.com to opendev.org so git-review works","message":"Change github.com to opendev.org so git-review works\n\nUnfortunately, git-review ignores .gitreview when repository\nwas cloned from an https: schema URL, when looking for the\ncommit-msg hook. It only obeys .gitreview when the repositiory\nwas cloned from a git: schema, basically ssh. It seems easier\nto change our documentation than to try to get git-review\nchanged. Also, there\u0027s an installed base of old git-review.\nSo, this fix changes a few lingering github.com references\nin places where the repository may be later used to contribute.\n\nCloses-Bug: #1947862\nChange-Id: I41ad14a08ab73f275cbde3abcf3938c0e5f3698f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d622770b5a96989d47fc521e163b0670eb6584b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d622770b5a96989d47fc521e163b0670eb6584b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
