)]}'
{"id":"openstack%2Fswift~814823","triplet_id":"openstack%2Fswift~master~Iacf66eda75fed6bf96900107250f393227c57ae5","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-03 19:44:37.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"Iacf66eda75fed6bf96900107250f393227c57ae5","subject":"Fix some imports for py310","status":"MERGED","created":"2021-10-20 19:30:19.000000000","updated":"2021-12-03 20:47:32.000000000","submitted":"2021-12-03 20:45:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814823","meta_rev_id":"23ff450b2426f722477e2f3fcf893006032514e7","_number":814823,"virtual_id_number":814823,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-03 20:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2021-11-26 05:05:03.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-03 19:44:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-20 20:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-20 21:55:44.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"dce1f9c1dbe27514b59ca72ef5dff77d4b7103c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-20 19:30:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41d575e3858ee4206a3afa455c4b783e402641f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:45:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e49376220b674810bd842f510f11d60c : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/4dc5c713ae4f492b825666c14fd36977 : SUCCESS in 1h 12m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6ebaf89731b4b869b726f495a5847fa : SUCCESS in 54m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef049bc3f258407a9d6dbea1676bb7ff : SUCCESS in 8m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9688bde58a8b470498bc7158e88e84b8 : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0fb2a682af8c4386aa85097f5f540c06 : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4aa3adcf667e4edeac82087e8a0b953d : SUCCESS in 16m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d12b31930e024eb4bd458b0a86622dc8 : SUCCESS in 17m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5fa7fffb85074d83a0c51983fe68dc0e : SUCCESS in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82c2ff81a6234b84a2cc100478ab7c2b : SUCCESS in 20m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c3f4630fbd64d80aecd1052a05833a7 : SUCCESS in 20m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7a251d7388c442f99532dbc96bd54d7a : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90f88379ef294ec39eab146d1b58172a : SUCCESS in 17m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/23be7933b9e04fd486fdfaa9a8d3d222 : FAILURE in 4m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/09e1707d5398437d9ab221c2c6b582a0 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/906cee29983546dab693c122e3e7f5cd : FAILURE in 3m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/48e89ed0566345ae88c649958088225f : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ffe3e6d35ee406c9b373b9b0e9b0aa5 : FAILURE in 5m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e97e6ca9cbc14769a8325a057264495e : SUCCESS in 1h 05m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/595a31948c3e4ea18b963501d27d4152 : SUCCESS in 53m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9cfd5bab8a7b440d87a8688b08441b90 : SUCCESS in 41m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e1aa4fbf2ed4d51ab256cb3bc9afe5b : FAILURE in 24m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32329a983e4545daa5948d4c078ba258 : SUCCESS in 16m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7883af3d56904801bd4f17c6cdc47fd1 : FAILURE in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/746248e2c3bd47ef9025d807eb15ccd2 : SUCCESS in 22m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b755663c4e64153b3f4cb38927cff29 : FAILURE in 18m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9c3381e73bfaa736ff54b93c77c606f5189f82c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-20 20:58:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31c877f521dc49c5befc0c5187b0da8e : SUCCESS in 31m 23s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa4b5efed08b452fb2ba9d61cb68f8f2 : SUCCESS in 31m 20s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/cd2315075a4f46a5a6bed3cb184bd8d8 : SUCCESS in 1h 20m 43s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8eb3b6900f8649a182a399fe3e9860e6 : SUCCESS in 30m 25s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b154251ab3b04415bfb8a31ce4569771 : SUCCESS in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cf13e37856c67ce2c456b03d78500ea13fb65d7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-10-20 21:55:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d406c9db238c1fefc5febfb80b55eb56f925e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-25 22:54:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"930ffbcef6d4a327f102f5343b1b8ae78dcaf27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 23:58:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab5351935bd449cb494dedee75148d0 : SUCCESS in 10m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/c6084530a2e24456911661c92411dc56 : SUCCESS in 1h 02m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c2f24bb13d441038e0b7f7fd9d5390d : SUCCESS in 40m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6540f6569fe14c099fb5fa4858abae88 : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/16cce7a2af4343e2bce6dd64e40a6286 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/556dd1bd8c2746a5bfdcc2285beb2d56 : SUCCESS in 15m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/66968ca9e8cf4d2ba03a4edcc885e4c4 : SUCCESS in 14m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1af09b1052c0470db3a4579b731ec73a : SUCCESS in 14m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/121ce2b5ffde4663a2e3331f447c2fb6 : SUCCESS in 14m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/273c7e9f0116401e86590a9b2b826527 : SUCCESS in 13m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8247f502b5a046f3a2d89975e7b7fec9 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/de15c3416f3e47b1aaa55dbf29c8b5a5 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f3c0a99bfb44ee0b5e078c3f01dc912 : SUCCESS in 17m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c85f64278c57410aa70b3e2ea9b1279b : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aaa7ffd93b3e44a8bf5ddb38b3763ffc : SUCCESS in 15m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6e967b9296524e0fbf89b8b257f0c9d1 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a207aacd434f44ffa573fcfb1f198290 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/611ae57ab9904afebf6f33a2239a1483 : SUCCESS in 19m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bcd7fcab56074126bd2553372172cf78 : SUCCESS in 54m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/caf2a4c6aa06431daa5022b3a44ada7b : SUCCESS in 52m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d07ed5a18dc247b7a2005d48cf22715a : SUCCESS in 41m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c44448f6939e4ec5a2b046e0f851b1e4 : SUCCESS in 47m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a3ada072eab43b3b23461294a5de90b : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4b6b9501624d5fae89c6ee7990ad6f : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa290a9dccb9408dbf8cbcf9ed50ea6b : SUCCESS in 21m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/398f7b893b2a48afbb8f511f1676a3d9 : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":2},{"id":"66714c1053feb8a3755bb13e14104ed66200e26f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 00:29:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a70a923ed3b244bf85c6e65cdebe3711 : SUCCESS in 31m 10s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70c90e28b3254321b7ab4022ad849181 : SUCCESS in 30m 54s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/9db5ae94e0c643e7bb60d8010c05b8c1 : SUCCESS in 1h 27m 12s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/02026259bdc24c32bcffa88a4ab354fe : SUCCESS in 33m 52s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/768ee574e3e1476ebbac70f29715ce7f : SUCCESS in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2cdc287cfe051fc5078dd6f8d6e673e691ddd689","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-11-26 05:05:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a842d0e2fa34bf4e6b46206d98b3417f70118f2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-03 19:44:37.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"124a420cbc592ae6445e3d5b63b262116db9fe08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 19:45:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc9147fb789ac22e3b881f1394a5162e3c946a15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:45:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddcf27007f5941eea1783cef1560a783 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/eb1054103aea45aaa27318d48558831b : SUCCESS in 58m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f5e40d9312641828d98e7c64003c922 : SUCCESS in 42m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd78429a9ceb42dfa7cd235a13d8049f : SUCCESS in 42m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/886bf106eb8544ada8829328fec0c854 : SUCCESS in 8m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b8d7a1dfd9dd42e28a4bc8cd2620fa2a : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26f9296f2b304f5bafb3f38108cdda95 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/08c5dbabe531496e9e721836ea719a00 : SUCCESS in 19m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4dec8c75bec44ed7b1ce9a696e24f60d : SUCCESS in 20m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0e000ac419504b6b92b3d5e4fb9754cf : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8fb0b49cc5fd442e983d9f25bbc7da11 : SUCCESS in 21m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b95d826ee42e4ca393571c2fba3fde27 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c381394a31884cc39df9168298480c1f : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd693be3c28344999541289958abf7c3 : SUCCESS in 20m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/973376bb955c4e53b691c35b10099923 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/59e911176c484eb0960db3751247055c : SUCCESS in 21m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5a98b30cf2c64f3995e057ad3a5265a5 : SUCCESS in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbec200fdb714a2c988663142d5f6417 : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eac79fa43a064d9d814d6eeb1170afee : SUCCESS in 59m 39s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f4aeb410fa774711a9b19d9fd0727cfd : SUCCESS in 52m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5089bae49ea42898312febc24bf4753 : SUCCESS in 38m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef30d51b141449a0bdd7be24bbf4e037 : SUCCESS in 38m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d622e08ba023489aa95e6ddcf28537eb : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7463455949cd4326b79c9ad7dd70e4d5 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb3ef76ce761401ab74c7ba021112e82 : SUCCESS in 25m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc5be4ef96964dbf9d148b4a7d2b2642 : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":2},{"id":"adbb8b9bfdd1966353b86969cfb04736ba6eaac6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:45:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"23ff450b2426f722477e2f3fcf893006032514e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 20:47:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2022247c044b9b939573f515ce0461 : SUCCESS in 1m 19s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/79b271495063494781d4119224bc407d : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2fb3e3e9b95e466e889eaa9fd6522fbd : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1eaf7474fe9b7d31f44de311011f85101a278445","revisions":{"70a9b9854230fe2025d6c6635fc0d4a3af2fb0fc":{"kind":"REWORK","_number":1,"created":"2021-10-20 19:30:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/814823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/814823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/814823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/814823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/814823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/814823/1"}}},"commit":{"parents":[{"commit":"a5fbe6ca41fbc66a0952f9430744fc3f552b335d","subject":"ec: Use replication network to get frags for reconstruction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5fbe6ca41fbc66a0952f9430744fc3f552b335d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-20 19:30:09.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-20 19:30:09.000000000","tz":-420},"subject":"Fix some imports for py310","message":"Fix some imports for py310\n\nChange-Id: Iacf66eda75fed6bf96900107250f393227c57ae5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70a9b9854230fe2025d6c6635fc0d4a3af2fb0fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70a9b9854230fe2025d6c6635fc0d4a3af2fb0fc"}]},"branch":"refs/heads/master"},"1eaf7474fe9b7d31f44de311011f85101a278445":{"kind":"REWORK","_number":2,"created":"2021-11-25 22:54:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/814823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/814823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/814823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/814823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/814823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/814823/2"}}},"commit":{"parents":[{"commit":"092d409c4bbe542f6b9f3c217aa05ffd8669c815","subject":"reconstructor: silence traceback when purging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d409c4bbe542f6b9f3c217aa05ffd8669c815"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-20 19:30:09.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-11-25 22:54:17.000000000","tz":-480},"subject":"Fix some imports for py310","message":"Fix some imports for py310\n\nBetween this and the (unreleased) pyeclib fix, I see unit and func tests\npassing on py310. Haven\u0027t tried probe tests, yet.\n\nChange-Id: Iacf66eda75fed6bf96900107250f393227c57ae5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1eaf7474fe9b7d31f44de311011f85101a278445"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1eaf7474fe9b7d31f44de311011f85101a278445"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
