)]}'
{"id":"openstack%2Fswift~815312","triplet_id":"openstack%2Fswift~master~Iee941343df08d01b9ad0c5c70fcf05240d259a4f","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2021-10-25 23:15:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee941343df08d01b9ad0c5c70fcf05240d259a4f","subject":"Test for PartitionLockTimeout","status":"MERGED","created":"2021-10-25 14:16:51.000000000","updated":"2021-10-25 23:17:31.000000000","submitted":"2021-10-25 23:15:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815312","meta_rev_id":"b0542af0adafbc3c15e8bff864111ce6f0c6b628","_number":815312,"virtual_id_number":815312,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-25 23:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-25 20:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-25 20:18:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 15:24:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-25 20:18:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a6fd2da452d9cbaa5c1f131a17d99b3d80f414a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-25 14:16:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c28753faf42427634624794a242e55a0608f058e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 15:24:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/018d268966aa409b8fb66d288dbeee75 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/1783a10a00d24717baf1801c77b69a7e : SUCCESS in 1h 01m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/abde9c26ef9b451c90bd78969e299128 : SUCCESS in 48m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aaf6b1e5726846e49697b0800cf1f651 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e547196cffe4f60b6fa9905893776e1 : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b023fdfa3144e9bbf3d70d947f0411c : SUCCESS in 14m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c257a74f780e4ca3869708a90ae6774c : FAILURE in 30m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/680d65e2dddd47149a6424b3fe87dc78 : FAILURE in 16m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/51134e9ac8f24dd388e765d67b094a87 : FAILURE in 17m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ccb49785f014507b8669a7b5b771949 : FAILURE in 19m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76f793cac58a44a6aeb005422a1e1f40 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c630cbe884234bbeb06878dcb5a1aca0 : SUCCESS in 22m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ddcb37a489c40ef95dc265372344530 : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6d43f7e7ae054711b25a61e1d30e8f3f : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/432b165e85d2410380717b0ada8acf4b : SUCCESS in 16m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/21f7be6c7d18438799a8cfec306c3413 : SUCCESS in 18m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e813e429c2a410ca9a538671ae5a8bb : SUCCESS in 7m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/29249ab4a61a4d719951404f493cc06b : SUCCESS in 20m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6875d1781a8c4a6085b35f9556f49b4b : SUCCESS in 49m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5d502d44beb54e6eaa16d326f5a1c90f : SUCCESS in 57m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d78fd5ec08b44956b14e996e8e7528f5 : SUCCESS in 51m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad8935473dfc4993a5032f716bf0bf0c : SUCCESS in 43m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/410c591558294571ae1b4f981ce770fc : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2862b47b39ea4ec3b95cede3f2a7051d : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45cd4189940c4c079f4dd1df34edae75 : SUCCESS in 29m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3052506780a46c3a7a18afc85309d69 : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":1},{"id":"14c4382d082810bbbd3642293c8b3ff20a3ea72c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-25 15:43:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8bd93ae38ad07c83e745595e290ad751cea1aa6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:49:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55192383be874e0399cc6a3e34b27eab : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/a1e50e084b1a49dd92117aa9706d0467 : SUCCESS in 56m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/927ad67baf8047f2ade5e1091784eacc : SUCCESS in 37m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a7630abd59aa4afb8b480995c7d3f7dc : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c9003b81cb734bb78be68a733983d5a8 : SUCCESS in 10m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0dada327aa1847859db9bcdafbec1428 : SUCCESS in 19m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6a4cb37b8164e3c83368d1468a87cd3 : SUCCESS in 15m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c28b8370a54476ab90b4364b05b7aa6 : SUCCESS in 17m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/af39b1050f3b4e34a252290a538312b5 : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb5eddb98e1143e09c96702501f85d74 : SUCCESS in 17m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/199f8f3493c04f569bf714d10e31c04c : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/55a88a74bd8648de9739d98e552aa4ab : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e4dce095944c4e96b6b50c6ee86b3415 : SUCCESS in 34m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1ee97e41d3834dcfb7f31bada000e0a1 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/893fc398d0a94ccc961daa8cb1bfea35 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ee33d02c247f4a5c987686ff08743f1f : SUCCESS in 18m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da848d44abb6414ca842cdcf151107a0 : SUCCESS in 9m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/31d0828f4de9490f836411a7c4a65c12 : SUCCESS in 23m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dfcfa17f28d040ebaf1e8850603af0ad : SUCCESS in 44m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3afa4393172245db9af937c10ecdeb4b : SUCCESS in 1h 00m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b92d4ba3acb44919d7d0ace6701717f : SUCCESS in 39m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a113db4b16241a59a6ea3315acb7e7e : SUCCESS in 40m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a7dbf6384bb4983b16583305d2fcb8b : SUCCESS in 30m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dea05fd47db4118bfb257014580858a : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/75048ab8220c4acea870ea9e29524530 : SUCCESS in 32m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ac9663ac4ae47399642cc9b4f899db1 : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":2},{"id":"7f08e05c1f23126f4e232f1e4445d8959a545e68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 17:14:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad8db9e8ce8e4d3fa335e9be5cbaa764 : SUCCESS in 33m 02s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e567bf38723e4655b4e291270b37fb94 : SUCCESS in 45m 27s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/03277036dca745db907dc18e5a016d15 : SUCCESS in 1h 21m 09s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e4f53c7bc6140de8a0b3cecfb2bb0c5 : SUCCESS in 32m 52s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/355e6575dd00477d9957d3d7b712d327 : SUCCESS in 39m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dfb9a9a79e664a6e9061f70d03e78596edd21d5d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-25 20:18:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d729b38300eb5b0a011ed212b41bbfb778bee55d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 20:19:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08c49d98579f4ce9d2a6a8caa020cbbcbd5aa0a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 23:15:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e0c497a6164db19d9562390394c057 : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/961859c02c9443a78f88f1a089380123 : SUCCESS in 57m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a0e32d687014df496886ca7d8c0f59f : SUCCESS in 46m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f924c88a33ae47c1ab8559a57abed223 : SUCCESS in 8m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/08a61cd03cd842f687fea3970b38df3c : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b9f8747c044441ddb70d064853eeb1f9 : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad860cb965e544b3b7d486698e2cbaae : SUCCESS in 16m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b2a886511f5e4d55a97f8d3f6ef4c81f : SUCCESS in 15m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3e7e10e484ca44c28c5b78b231731370 : SUCCESS in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/91c21823d7b34317a318b09e5945482b : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ef77fe1226cc4c47b211f0294ee58ad9 : SUCCESS in 1h 05m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f953503dbb954ba3a62ea99fb5016a89 : SUCCESS in 27m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f16035b48404cff8b9297791f7e9041 : SUCCESS in 20m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/daef879307754e48babec75a6ac19bf9 : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21013dd3b9a24256b6338325c028fa53 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a513f437c9464330a03f3fa3898e8c94 : SUCCESS in 21m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01eee857929342ecb499ce598bec0680 : SUCCESS in 8m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7f7b4b8b03140e380a69fa8adbf0ae2 : SUCCESS in 49m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3909a105ef9b4c0b8c11dde51036ccfe : SUCCESS in 1h 01m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2713df02fa7f4caabcb3f17ab730cfd0 : SUCCESS in 46m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b4fd4cc462f48909d507632064f85a7 : SUCCESS in 43m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e22016a36c684cfe92bf2695659ada15 : SUCCESS in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0ad559873c4d30979f87921883f87e : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a22478cdd1304e20b7930db8e7f39eb7 : SUCCESS in 23m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6106434191744f08f548c9c580cdc7c : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a170998b511d3eddd5e29cf8d8bfdf88b28d08b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 23:15:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b0542af0adafbc3c15e8bff864111ce6f0c6b628","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 23:17:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db2ab0a4475444cbb046a79717a1c8e6 : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/31cdafa0f9cd4ae3a026d3747d6d09b8 : SUCCESS in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/da4e368757b44a5fbf5424fc9c389ec1 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ebd0354a9547fef1d6a08d1d3c41a04e2bb5cb7","revisions":{"39e9bf473cb7435ff85f6b5b3ae77a9f978a69bf":{"kind":"REWORK","_number":1,"created":"2021-10-25 14:16:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/815312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/815312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/815312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/815312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/815312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/815312/1"}}},"commit":{"parents":[{"commit":"336599842481f449833515ea96b3c616f4ae39c5","subject":"Fix log message interpolation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/336599842481f449833515ea96b3c616f4ae39c5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-25 14:12:54.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-25 14:15:17.000000000","tz":-300},"subject":"Test for PartitionLockTimeout","message":"Test for PartitionLockTimeout\n\nRelated-Change-Id: Ibe139a4d303079b5b803a76b270e1366f09dc534\nChange-Id: Iee941343df08d01b9ad0c5c70fcf05240d259a4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39e9bf473cb7435ff85f6b5b3ae77a9f978a69bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39e9bf473cb7435ff85f6b5b3ae77a9f978a69bf"}]},"branch":"refs/heads/master"},"8ebd0354a9547fef1d6a08d1d3c41a04e2bb5cb7":{"kind":"REWORK","_number":2,"created":"2021-10-25 15:43:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/12/815312/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/815312/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/815312/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/815312/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/815312/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/815312/2"}}},"commit":{"parents":[{"commit":"336599842481f449833515ea96b3c616f4ae39c5","subject":"Fix log message interpolation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/336599842481f449833515ea96b3c616f4ae39c5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-25 14:12:54.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-25 15:43:18.000000000","tz":-300},"subject":"Test for PartitionLockTimeout","message":"Test for PartitionLockTimeout\n\nRelated-Change-Id: Ibe139a4d303079b5b803a76b270e1366f09dc534\nChange-Id: Iee941343df08d01b9ad0c5c70fcf05240d259a4f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ebd0354a9547fef1d6a08d1d3c41a04e2bb5cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ebd0354a9547fef1d6a08d1d3c41a04e2bb5cb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
