)]}'
{"id":"openstack%2Fswift~815330","triplet_id":"openstack%2Fswift~master~Ida23f61d7874f34b340b24e02f754eb955e41c82","project":"openstack/swift","branch":"master","topic":"primary-only","hashtags":[],"change_id":"Ida23f61d7874f34b340b24e02f754eb955e41c82","subject":"obj-reconstructor: add primary_only conf option","status":"NEW","created":"2021-10-25 15:37:52.000000000","updated":"2021-10-25 17:14:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3a5784e72b199ab13b44a570f68cfe30f0a762e9","_number":815330,"virtual_id_number":815330,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-25 16:45:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-25 16:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d64dac999b41e8230097d8be09060d2792e9f2dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-25 15:37:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"970b3353f0ebc65b9dc746cb45f9e589cfd1ae08","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-25 15:40:39.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b4bdee5871bf944d9c4b94602153c26cca113d0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 16:45:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb179da6c7224090bc6cd257a39cf06d : SUCCESS in 11m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/d0009d58a1ee45328b195bc518f84a9e : SUCCESS in 1h 01m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dcae33d54eef4d8596c91687e587569e : SUCCESS in 47m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c2c7cfdc10f5429f961731675ca2dbd1 : SUCCESS in 9m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/feca9cd9088d40a29e1d619164cdf241 : SUCCESS in 11m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45f15b08b717413e8252c86cbca38502 : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab19107cb60c4af297ca544be2da78d1 : SUCCESS in 16m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/16b8a7bd0a6f48f7bb244fab2b6b028c : SUCCESS in 18m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/750e738cd5844c999c7a60964f94accb : SUCCESS in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c9382ee988444118f51a045fda4f85c : SUCCESS in 17m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/04111858354f4209b7641158780085b4 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc86fba61e024b599a62783feaf789a0 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/75ca3052906f4aa0b30fe7f1de2d258a : SUCCESS in 19m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7a97ad9991aa49efbe26a5b4fdeb4f2a : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5b1b1b53d8a04c7891504f142ab27752 : SUCCESS in 19m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9076e7576e044377bff5001d577517de : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3836a1fcf3bf43c8b5b010198249da42 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc4687795a004cab88818d7ae0ced3d6 : SUCCESS in 20m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6100cbd51ec473c9f107c6c2b435bf1 : SUCCESS in 52m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/483d84ee76d441d9944d0a3e906096f3 : SUCCESS in 59m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/564a6b6a54cf4c1895348ac6f2b3f6b8 : SUCCESS in 38m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca107f9ca1b44be3bcb7e1a9d89b4e72 : SUCCESS in 42m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb0e6b5cac4a458b91e17da7b916a94d : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606c5b7631ba4ab89222901029209879 : SUCCESS in 6m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db6f2a3468a34d1c9acac1c068734300 : SUCCESS in 25m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/289f345cea5f4d4da7d22c7ef9eb38db : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3a5784e72b199ab13b44a570f68cfe30f0a762e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 17:14:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99f7d1349e224eae80af93e409092c0c : SUCCESS in 37m 18s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c05363157e1f4280a04cc7ee31c95c92 : SUCCESS in 38m 26s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/b57767b24db440eea7af66bfc03af567 : SUCCESS in 1h 27m 06s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0ff3a81ac71458f84e5e2bce1ae6254 : SUCCESS in 36m 36s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9897f3a17b384a8ba0df4f9212e8ef86 : SUCCESS in 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7115f8eb40adfb12a8ce702af2435d384d19c62","revisions":{"e7115f8eb40adfb12a8ce702af2435d384d19c62":{"kind":"REWORK","_number":1,"created":"2021-10-25 15:37:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/30/815330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/815330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/815330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/815330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/815330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/815330/1"}}},"commit":{"parents":[{"commit":"7f5e34a405822773b8c1422aad4ff64b5f645d14","subject":"sharidng: update doc to only mention auto_shard experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7f5e34a405822773b8c1422aad4ff64b5f645d14"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-19 18:50:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-10-25 15:37:14.000000000","tz":-300},"subject":"obj-reconstructor: add primary_only conf option","message":"obj-reconstructor: add primary_only conf option\n\nChange-Id: Ida23f61d7874f34b340b24e02f754eb955e41c82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7115f8eb40adfb12a8ce702af2435d384d19c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7115f8eb40adfb12a8ce702af2435d384d19c62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
