)]}'
{"id":"openstack%2Fswift~815520","triplet_id":"openstack%2Fswift~master~I51a8c1fd7a179c1a53637f8f22c2009cc404e8ed","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2021-10-26 17:31:09.000000000","reason":"Tim Burke replied on the change"}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-10-26 17:31:09.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I51a8c1fd7a179c1a53637f8f22c2009cc404e8ed","subject":"Ensure update_metadata() argument types are as expected","status":"NEW","created":"2021-10-26 15:54:19.000000000","updated":"2021-10-26 17:31:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d43d19f420569bd1d0fb50b90b665c7cd64bd7d8","_number":815520,"virtual_id_number":815520,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-26 17:01:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 16:48:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-10-26 17:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17ba1f093373db4453ed93d6de6543827293fd4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-26 15:54:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19ef7c6d52d05363aacac729e753b91b87a5d1f1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-10-26 16:48:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a45a74bf2d14f9fb1fc5939de0fa521865990f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 17:01:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8b56015cf940a5851bc5d9f2b35223 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/0efa45b1bb2244c28913dffdccee9b28 : SUCCESS in 1h 02m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0694f98d69154ec6b42fd8a1be61fcdb : SUCCESS in 43m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/95b3caa2e3684bf08272315127e9806e : SUCCESS in 9m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4547e4cdc56c478d95b7fa86800373d7 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0560c308c2de4ee7b5d800dc4f10ed62 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/efe90e40898a4b1fa9c14c0acd8c6d60 : SUCCESS in 22m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c0e757d9fb5f480bb363bd3971b773d8 : SUCCESS in 20m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a62a100944854d87afd94fb6e0948902 : SUCCESS in 21m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7471837f5deb4c3cb8b6e788c054bb55 : SUCCESS in 21m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c965fd046a245e481f2ca59f98a01e2 : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e950ebe3809f4c81a378cacb7a9c6c34 : SUCCESS in 21m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/142eafefb3e54a5a8269fd579c21dc16 : SUCCESS in 23m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6851ee71df8a4127ba3097467f44556c : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f2edc3cb233b440a9787ffb3c45fd1b3 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/42710ce776a54c3bae5fa7f6269316f2 : SUCCESS in 20m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/df00254140534233b70d4cd7351d60a1 : SUCCESS in 9m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd81dbecc1374a539fc6bc8acf886f76 : SUCCESS in 27m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ae0e590363c448ba87227c32b6b4cf1f : SUCCESS in 57m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/479be0742e26497a92e84fd720b7f477 : SUCCESS in 58m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d97a28ef1af4bd0b2b9bc1e12416776 : SUCCESS in 40m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab6a90395c2b4a24b0ff53beb1e30a42 : SUCCESS in 45m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b98f3c749c884bb3b9633d0bb2cb6f94 : SUCCESS in 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/571b563cae6d4e979632fb35f9828f9c : SUCCESS in 9m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/44a2c1a58f9140c4a5eed8c74f351108 : SUCCESS in 32m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a07d28c5e7d24a05aa438cb06df34481 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d7b32006bd8fca91dd678084474068e8b78d769f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 17:22:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b4715ca8873417cbcad5102e4b894dd : SUCCESS in 32m 55s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/950e2aeb32624274bc11dbbb337bb6d1 : SUCCESS in 33m 03s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/4ba1200d29064e93982f9ef0bc28ba91 : SUCCESS in 1h 21m 09s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fcefee60fea4392a74c8c0d6828d77d : SUCCESS in 31m 52s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bde9fa06fe104b48a0e6b851e4e450c2 : SUCCESS in 55m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d43d19f420569bd1d0fb50b90b665c7cd64bd7d8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-10-26 17:31:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acde736ce2a1dbfdfbd0e0df24b704d0f0107318","revisions":{"acde736ce2a1dbfdfbd0e0df24b704d0f0107318":{"kind":"REWORK","_number":1,"created":"2021-10-26 15:54:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/20/815520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/815520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/815520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/815520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/815520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/815520/1"}}},"commit":{"parents":[{"commit":"fe8344234004dcfbede5708f7fbd09c677fa474b","subject":"Merge \"Test for PartitionLockTimeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe8344234004dcfbede5708f7fbd09c677fa474b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-26 15:53:53.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-10-26 15:53:53.000000000","tz":-420},"subject":"Ensure update_metadata() argument types are as expected","message":"Ensure update_metadata() argument types are as expected\n\nChange-Id: I51a8c1fd7a179c1a53637f8f22c2009cc404e8ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acde736ce2a1dbfdfbd0e0df24b704d0f0107318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acde736ce2a1dbfdfbd0e0df24b704d0f0107318"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
