)]}'
{"id":"openstack%2Fswift~816119","triplet_id":"openstack%2Fswift~master~Ibb2cfa6e324de0801468860ab9a71edb94ef5c99","project":"openstack/swift","branch":"master","topic":"middleware_trace","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2021-11-08 06:10:27.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ibb2cfa6e324de0801468860ab9a71edb94ef5c99","subject":"trace: Add RequestTraceMiddleware","status":"NEW","created":"2021-11-01 05:10:40.000000000","updated":"2022-09-12 02:01:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fbab2d4317a34956f847f8ea0cdf511dbd7d2705","_number":816119,"virtual_id_number":816119,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-12 02:01:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-01 06:17:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49e3baae4fc1b3e060ae1cc281b29fb5d50017de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-01 05:10:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f51e84f7984a43d21a12eca0663cc1ee25eee51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 06:17:56.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208809da270648ad825ce7132250ac6a : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/79ee9b26d60b41398070e09e8473afbe : FAILURE in 19m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a115c5246d44d3bb068b24946b0737d : FAILURE in 38m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/195b677ac818464ab070b346e081f82a : SUCCESS in 8m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2d5611800dc448b49b1b1bd6d6b3b3cd : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c5a7dd15b330442f83821f7407e6382a : FAILURE in 18m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67d93c3c22774074b464ccc627fc5640 : FAILURE in 16m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c03990a51b48464589ed92f4320d538c : FAILURE in 16m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b56eb38f1f114a85908794502d1930fa : FAILURE in 18m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b30f7d01c8945a9872c01bc2283eabd : FAILURE in 15m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6b4d996c0c774c7eb0e88324ce1d1a57 : FAILURE in 22m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94a11709b7aa41bd8ad5050c7efade4b : FAILURE in 18m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffe0d1fba11643598c296bb03c1a121b : FAILURE in 23m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c497fbd6404f4cd3ae68af986a57d807 : FAILURE in 16m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d56994f8ba16437f889f16b158edc5e8 : FAILURE in 19m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8b0bddef47634493ab03f282bb65ed96 : FAILURE in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c67f7f4cef7041518b9dcfef64c340de : FAILURE in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8406757d23f4af780e394ee9506883b : FAILURE in 17m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1eeb6bf14e924f5e8661838e59b31dbb : SUCCESS in 58m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/71b20eaa27074f29931c96c8b644b694 : SUCCESS in 1h 01m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9183cd33985f48d2abd5bc8ae25bf40c : FAILURE in 31m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f23fe55df1d34566a7fdb29c71b6e15d : FAILURE in 29m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1fa7f5ddf824d0b90337f64e9f3a387 : FAILURE in 25m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2439d42e72f642398b158a5d9075bafa : FAILURE in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47098de631514db4b25e5cc3402c0422 : FAILURE in 28m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da372c91a4d94b05819800c130fae18d : SUCCESS in 51m 11s\n\nWarning:\n  Comments left for invalid file swift/common/utils.py\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py","accounts_in_message":[],"_revision_number":1},{"id":"6c32b59977c3adf18dc9f6b4ff96f6fa80acb432","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 07:15:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0781bb0f0b6b4264b67ed781d14666e7 : FAILURE in 41m 47s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b7cf2030c344e6787ec0cc888e099a7 : FAILURE in 47m 05s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/14469c54390f422ba41b5158359ce1e7 : FAILURE in 2h 01m 58s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f95db8977cc44c7d9a7353c2efafde25 : FAILURE in 44m 10s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/234f0258787c4ab08fe5f08b15f5e567 : FAILURE in 43m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2945dde110f8acfba766ecdd767f3f3b70b479d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-08 04:30:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56b3dbc855b1673d2b47162ebd8c22579661edb7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-08 04:35:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"475480ddd313ce73196376dce0aec43e68c0fb7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 06:04:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b5f394f56b84656a7383e3299eb456f : FAILURE in 26m 35s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae4c9735f01f4ade8947e915901c342b : FAILURE in 29m 25s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/b8bc8854e074436b9ae056b9b397d5fa : FAILURE in 50m 50s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b27e39e95c454c4ea740b2f2acdc7ed2 : TIMED_OUT in 1h 31m 12s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7467ee00baf040bd80167834c6f31f62 : TIMED_OUT in 1h 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84b28570ec683b17769799f9c18bc57454ef3f37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 06:10:27.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbe78fade624e65bd96bb18c4111e08 : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/460dbf7f743a42cfab2bdfc9d7fc58d4 : FAILURE in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8ca620a2fc5244d69d9f6f3bf6e932a4 : FAILURE in 34m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a09034a2fe0a449db2b0f685b76e43e2 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3bb701090b374f73988ea6a8a8a3e187 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e18c6d074784ba790e1178d17ae963e : FAILURE in 17m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d1fb1414f2704d1e98349aa5e63a7b30 : FAILURE in 16m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7d3b9df859e4034b3ea71b493982438 : FAILURE in 13m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aa0a835c6ebe41598152d818886115e7 : FAILURE in 13m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0ef8f2b00dcd4a89b721ba07ea86f46b : FAILURE in 16m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4ab4d3383ca84fcd96bc309bbb2ad163 : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b9e7142b6d845b3849556babe9716ab : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c324f91367fb48a08b169797bacf233b : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/353274b7e08f45479c8a7847f6fe518f : TIMED_OUT in 1h 31m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/425edbb4fb494aeaa32b4dda9019f399 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b389d446926d4aa0897ce7d275e0a09a : TIMED_OUT in 1h 31m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a392823c49c94265835c931b07763d51 : FAILURE in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12a24d93ad564415bae2ba13ca1454ee : TIMED_OUT in 1h 31m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cc5def463eec4c468453d3780290244d : FAILURE in 35m 54s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/90ef50db5b59441090f925fde752ebda : FAILURE in 29m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ab1b43ac6e04bb8af3f465b22e13086 : TIMED_OUT in 1h 33m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dce28801ee994eb7b2d9783b55c0d624 : TIMED_OUT in 1h 32m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20a1b9685a564972b9b378fabcbad29a : FAILURE in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c66d8c5bbf3d46a5ad59b3c9e7c20183 : FAILURE in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a51f2e601470481b99c233055a331aa8 : SUCCESS in 24m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d8c5d2569b24422a60992fae7f7ee4f : FAILURE in 31m 45s\n\nWarning:\n  Comments left for invalid file swift/common/utils.py\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py","accounts_in_message":[],"_revision_number":2},{"id":"a876e926963747a7f942cfbc182f74a5f0283b1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-11 03:37:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9f804190868bdd84f91ebbcc14b2fc84d1c1a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 04:48:18.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83358b4c31f5487999e6621dc70362e8 : SUCCESS in 11m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/f6d92090c6c74811b1839d58cc6357ec : SUCCESS in 1h 02m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ab35bc252d64cc3926955e1220b2ff8 : FAILURE in 42m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eb9dc6073d004d85a16275d053d5a94a : SUCCESS in 8m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/db53e333ca1a475dbf425bd788c2b306 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/353430b46a4c49418d106c5bb72d7150 : FAILURE in 16m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d2f9f8a40d9e460fa920511888508679 : FAILURE in 16m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0536812456c846caa36c511a182948b7 : FAILURE in 15m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b96c2996350e4b559be6509ca83aee89 : FAILURE in 16m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/741167b7b4274551ae16055f05cc7504 : FAILURE in 20m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa814323cecc4ac19b027496d744b8c6 : FAILURE in 20m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0154339ec123451f919abb42bc95c3d9 : FAILURE in 21m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15ba4af5484f416f88128b62377aaceb : FAILURE in 21m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2aa4044f96c44e2a6528e975a0def3f : FAILURE in 20m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1937c99a021843809b2cd0eb0aad5965 : FAILURE in 20m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7c99e72aef91496e91880552f0322204 : FAILURE in 18m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4a1409b491248e190fe759b1cdd30e3 : FAILURE in 7m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/affd4426724a4ebf898de1bdc876daff : SUCCESS in 25m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/96cf3c67a691450785b6abafa6c6d12b : SUCCESS in 1h 06m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d80b615cb8a845c6b067400e34050864 : SUCCESS in 1h 05m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbafb30622ae48d3b028c216eecde3ae : FAILURE in 29m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7331c29a0bcb443895da12ebe00ff301 : FAILURE in 33m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a401aff57ecf4f5e9dbd600b8ff741cd : FAILURE in 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4e496d4feb44ce898110627480f1e2 : FAILURE in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce8295f5c9284a7880792b0273d655f5 : SUCCESS in 26m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae47cd61b79e47308dc83cb0eb74fed0 : SUCCESS in 40m 31s\n\nWarning:\n  Comments left for invalid file swift/common/utils.py\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py","accounts_in_message":[],"_revision_number":3},{"id":"20381acf4cbadcdcac332bc8232e7b0e1f403de0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 05:23:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d77c6f750e964f9485d243a184e1e550 : FAILURE in 37m 49s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/942f3b49f1514efa9ba3b0841c4fa41e : FAILURE in 38m 10s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/8a77053d4cc74771b508e157d6ce2844 : FAILURE in 1h 38m 29s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fc216f859f24c97bcf9181a1875181e : FAILURE in 39m 47s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/be52ce81d9f642eaae984fd7dc624cec : FAILURE in 38m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e212085cd8eb5d5d0e9e46d0dc59b6b56d7c2fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-16 03:50:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67348290d4939ed609e17b4d4868d334ba2d0e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 04:50:57.000000000","message":"Patch Set 4: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa67ac125e2548b38c3825ee0d2944a0 : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/4899d3d0988f4987900190ecdccc3b71 : SUCCESS in 57m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1b465c5334a496b9f0f6ae6826df67a : FAILURE in 35m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce4b534d77894c35816829854b04a296 : SUCCESS in 48m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/340d956f21c549cf90879aebc3fcd0c0 : SUCCESS in 9m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/66613fd12050426593c00afb14d5e85e : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a2df6500ef94e609c57c367713c43a6 : FAILURE in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac64cc5704cc4357bdbd6a76a9492442 : FAILURE in 17m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d270d252d084298bcc9a41b649c9653 : FAILURE in 16m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d81611cb95e2424685f802f417704fe9 : FAILURE in 15m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8fa7142c166a4a08b2e0beeffd0afdd6 : FAILURE in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c7739a4ec0b4830a9328a7176aaad87 : FAILURE in 27m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/365d7cbebf6c4b07819439318bd5db79 : FAILURE in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/859ce29a60e64c0aa08903c0b4e1bdf6 : FAILURE in 19m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7324aa7208034eaf993fb04c4ed45344 : FAILURE in 27m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f0da6f0454eb407c960dcf3fa580f919 : FAILURE in 19m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d794fb48250d4b11926a05527b12aaa3 : FAILURE in 16m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/869beb21a6f64aeabaabb36fa263889b : FAILURE in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2dfd77d63f8401baf9a16c84b82b0c1 : SUCCESS in 19m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9be1bd0697eb46b7a22a1b3a4c7e35b6 : SUCCESS in 50m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ad4a621178d4ba5af9e70fcdf1f3904 : SUCCESS in 53m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2505b787b4bc41cc8690640c335fdb3a : FAILURE in 32m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9da739f1ae46416e863986fa3c59168b : FAILURE in 30m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc125d76d8564e1fb0bdfe9733f4492c : FAILURE in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed17b5eb7e04995b49de9ba98fde58c : FAILURE in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92af1c7ec2834e218cae5e0988f74ad2 : SUCCESS in 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c906b6ddb9904ada81558b1ba52d62d1 : SUCCESS in 51m 27s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py\n  Comments left for invalid file swift/common/utils.py","accounts_in_message":[],"_revision_number":4},{"id":"e15e03f65bdd2a173a4164507d48a6878706eb84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 06:08:51.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0e564a551e646ff9258dc74000b8330 : FAILURE in 41m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d780a5a1bded4e04bc29f5baa937791f : FAILURE in 40m 26s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/04858ba1d97f422db809ae482f266032 : FAILURE in 2h 04m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8eeb5531fa6f43dabb0788f16e559d92 : FAILURE in 43m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/da39eed6b1ac496187148943d2aa2062 : FAILURE in 1h 07m 32s","accounts_in_message":[],"_revision_number":4},{"id":"e6a9fd54046f76c8a7470598b36c7bac491326ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-11 01:31:05.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a08871470f5dce23815f8e8bdc1dd0589b6abe18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 02:38:49.000000000","message":"Patch Set 5: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d349bd78cbc1489e9201e92da13ce23b : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/9b78e7cf253e4387b64c1691ef219aab : SUCCESS in 54m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f637b7f814f4421881e9c4627d937d29 : FAILURE in 45m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f821b75284cd43dfafab3a781f255fcb : SUCCESS in 44m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/828995124ee1406bad936c0114bfb101 : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/74f4bc3cb55e47e088f8275b5e22d464 : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df59e3ff81be4571b80d5c4653f13317 : FAILURE in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8796be18c214ad8bb546d69e91c11f7 : FAILURE in 15m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/507328b3a79e47559e053ed8e469a1c1 : FAILURE in 14m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa7e4e097e58413f93801f01fb56f2b1 : FAILURE in 18m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa56da57f9eb492081edfbd5f48ff01d : FAILURE in 21m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/35be98509a6f43f6aa33521bae7ffd4e : FAILURE in 29m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4064583e33e14a55a34b9d83e2a8caa2 : FAILURE in 24m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/573a3c86cd9a4415a25629a32dbaa012 : FAILURE in 18m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/80233bf9605948aeb722ea113ad2b743 : FAILURE in 21m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23b538a29a4548b99d7c31894d29c1f2 : FAILURE in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b777ac40dcf94c72bdc28383cf6a0850 : SUCCESS in 20m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c69fcc0a69aa4c2cb83c37c9d66613f4 : FAILURE in 57m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/40533767655348a388c2d67988831767 : FAILURE in 1h 02m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e742c907201143668456069a538248bc : FAILURE in 51m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6e838fb68624ad09a50480b3e690ceb : FAILURE in 54m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e168e07104c74740a079a99ec3c2f392 : FAILURE in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad65f0aa0e84da8ba85cbf239719149 : FAILURE in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/224e48d210b441c69d0dccfa6d047d21 : SUCCESS in 32m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a140c8eee1b4bb09b26884577895213 : SUCCESS in 48m 44s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py\n  Comments left for invalid file swift/common/utils.py","accounts_in_message":[],"_revision_number":5},{"id":"89423869dc94df7fa4802094bcd01daf75d5e517","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 04:07:50.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d127e191a5c24046886e7ce18b7018eb : FAILURE in 39m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27ef5f82b6f446f3a3dfc2bd4ec189b1 : FAILURE in 40m 31s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8740e6bc38c344799e3565b2b80f78a2 : FAILURE in 2h 33m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e71f87f0f4464251a6894e8945acb959 : FAILURE in 1h 00m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2243f1515ca4da893da98c991c56bff : FAILURE in 1h 13m 08s","accounts_in_message":[],"_revision_number":5},{"id":"00c5f765d2eb940ae483012b9ee36985a639b7c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-31 02:57:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3afdb0d5a105c0cea6b15fe466b239bbcd1c13c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 04:05:44.000000000","message":"Patch Set 6: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f92a94295934cbfba39bdd9987fc1f3 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b13ccedcbb4d539ef290fa8abe2922 : SUCCESS in 10m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6740a6eda6b040a689567edf34864cd5 : FAILURE in 3m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/dec0102be4124fdeb994b1863576765d : SUCCESS in 46m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b776be122414ed1916bf520b695cd02 : FAILURE in 1h 01m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17bfb7973e384bdbbef2b01816f5fae8 : SUCCESS in 37m 49s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9c9945ac8d8a45929fa603a999121470 : FAILURE in 27m 06s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e4d18decc34446829fcbe1094e0edd19 : FAILURE in 27m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1a5d689b22b34ec39bb80989f200ba23 : FAILURE in 27m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3df561d7be5e4994961f9c7431cfba6f : SUCCESS in 11m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d6ec6106347b46508172c48613fd49ed : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b18466850ca144fa837b81a7f2ec5a53 : FAILURE in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc2923f41e2c40e189b1e5bc28a7562b : FAILURE in 11m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c936eefcbcd3448d82236a073c1bf5af : FAILURE in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92d9973cb6d04733b1b475447c77caef : FAILURE in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b124afc5ce2a49caa77c90bbfcfb0aea : FAILURE in 17m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3fac8348c7dd457ead3e029543e4c4d0 : FAILURE in 25m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/919a457d15fe44f5a5fb08913a3165a0 : FAILURE in 25m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1ae185b91c647e39d615ecc3e9e5bc4 : FAILURE in 20m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/14543bf4c1b34631a44f8d0983ed3d15 : FAILURE in 29m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7feeeb029ac245ac863f2ca34c84929b : FAILURE in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/07f552b4bf5a4db695f983404ee04a68 : SUCCESS in 24m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1d294430a15746c7b144935397f6156a : FAILURE in 1h 06m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c427fd7569df4fda96a6fc2a76570afa : FAILURE in 57m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be8aba720baf404abbbf16d6c27f58fc : FAILURE in 52m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/daf9ceeb0aaf4080aed6ae2dbcfb70cd : FAILURE in 45m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e35a09bc521841e9b5a146d722a332da : FAILURE in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14838b9ad9fe4de08c6fa26ebe625370 : FAILURE in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48795cca7df145e78b47d463d9c129e9 : FAILURE in 31m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1828ea93e7148f88fdbb22a2f5b5f04 : SUCCESS in 50m 00s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py","accounts_in_message":[],"_revision_number":6},{"id":"6cf1890c2fc2ab0203ee81f1fbc1c521208eb4f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 06:06:46.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b9393381dd5477b87e57bf18a1d354a : FAILURE in 29m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c40cfc1f39fa43ffa7137388969b76c0 : FAILURE in 38m 01s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/187020fbfd0649bbb762ac4ad9a8af99 : TIMED_OUT in 3h 01m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d90f4503fe484f2dab8c6af96dec1661 : FAILURE in 1h 19m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6012c4b1a50435385a6d33d06114cc7 : FAILURE in 41m 20s","accounts_in_message":[],"_revision_number":6},{"id":"94e3be04ab219e7df0bcde09e81f1ea28cf93a2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-03-09 22:37:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4da67a29e30df36a11ce2b59891a27dc427ed2dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 00:17:15.000000000","message":"Patch Set 7: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d39742729427498ea483499330695bfc : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8af7ab7f4994fdd9fc0f974d5d7d35a : SUCCESS in 9m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb60a061373e4904a92426aab6865636 : FAILURE in 5m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/af749a5caaa44ccc9f24a2559dbde4a9 : SUCCESS in 45m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a97913da14474388a9aeb39d3b53aca9 : FAILURE in 53m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d039c48c1ce14fdc8869f42816ed469c : SUCCESS in 48m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/22ce0878a7d94eba912f3d28ee5d64e8 : FAILURE in 27m 13s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9e0dc23cbba74ae39ad60ca8c4fa218f : FAILURE in 28m 05s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eda63d9ae2494929bfd0893cc312e09e : FAILURE in 29m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83447f3cedb04c55847ee515274d714d : SUCCESS in 11m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8b42b9d4b3f24d1f99bfc4fdd877e4a2 : SUCCESS in 8m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b8d3585701541738a36e007c47b0865 : FAILURE in 19m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/77af267e454249128d5581236f06bab9 : FAILURE in 27m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a78058d60b124b9f8b2f8b179646c2c8 : TIMED_OUT in 1h 32m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/183e382a8b5b4e77a70b70880285a7e5 : FAILURE in 22m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c383030f1f03475b89b849410c74dc70 : FAILURE in 27m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a36b535e5b064da6b91b1fff5251d0a0 : FAILURE in 26m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e7bd6dac741e435e9357ae2237ddaacf : FAILURE in 29m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/92340027306a4ae69914f37ce3e40890 : FAILURE in 27m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2ee7069d806f48baa5d568c28bf0fdea : FAILURE in 20m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/09edef13469c4d86bda312df8fd26726 : FAILURE in 9m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff1b84106add4da38099cc6b038d6c8c : SUCCESS in 26m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/18f4bec1b4cc4be787c8d04f90c7338e : FAILURE in 1h 02m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d3724ecc3d54eb6b28c44cb415a1e4c : FAILURE in 1h 09m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c856d3d39bd4dafb620902e0d95499d : FAILURE in 49m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf0561b7457d4ca7aafffa3da342715f : FAILURE in 54m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49babf3f734643398b4b570d6df16d70 : FAILURE in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7037e776638044259125edc87eca16cc : FAILURE in 7m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e2c6630e84714a7c990ea74153a7c0df : FAILURE in 24m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/161d735b042e4afb932d5109075c484b : SUCCESS in 43m 31s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py","accounts_in_message":[],"_revision_number":7},{"id":"15856903bcbfc5d365296d037b3437aae2ef8054","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 01:43:30.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4640319184148d689293420d670972d : FAILURE in 50m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20086c68f4a445ddae116d5d7d3f3a3c : FAILURE in 52m 23s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/fb24e1fa1280444281eec3907ea128b5 : TIMED_OUT in 3h 02m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f582d80204bf469e82268af7f57db1f5 : FAILURE in 1h 00m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fba0643e6384373ba9b581fe55352d4 : FAILURE in 1h 03m 27s","accounts_in_message":[],"_revision_number":7},{"id":"7d6973025cb5c2c6b00afe4ae6ae957bb65b7f6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-09-12 00:31:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"97cbf0e7527fed9459da5c98b01b34f6744440ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 01:29:02.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/394a7b3e67734e719af5a328365ff050\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7757407ece634728aae064db26329f5c : FAILURE in 50m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cdcab26ee2b14427b3b8dfebe36ea987 : FAILURE in 29m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9adef7eed834608b6578839e9c8b59e : FAILURE in 54m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/af2c9ca8e30d4e1c93a0e69a8de51093 : FAILURE in 37m 44s","accounts_in_message":[],"_revision_number":8},{"id":"fbab2d4317a34956f847f8ea0cdf511dbd7d2705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 02:01:39.000000000","message":"Patch Set 8: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0813439703b4ac8b651f2c612a9dd7a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c1929b065b84e84896886e7e38d91da : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aa6ad7a1816459eab47c5b9927c1d0e : SUCCESS in 10m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/26579dae9be14904ad4114a3c9f1e17f : FAILURE in 5m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/7fb54355a0a14d79a15683cb8b5aa004 : SUCCESS in 46m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1149e3840bd54005b285d33bc5912d7b : FAILURE in 31m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be87d333ce33428197e049cc4381d91b : SUCCESS in 1h 24m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e81fb074c3f45e09be3c833647e35ac : FAILURE in 26m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e52ecf643e2243619dbc12ff593d1ec3 : FAILURE in 26m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d692ab304a0e45fe93b2b7b02a3b5bd6 : FAILURE in 24m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/61eeee4bef044e4c9825080c2508137b : SUCCESS in 8m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ae18036b61d242928e775ecab7d901a3 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3d330b6cf2424474a3f35dfc7389416f : FAILURE in 14m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7494bbf8c42478497ec20a2fffa3f68 : FAILURE in 13m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/50227032eff340159153c1735f78ce5c : FAILURE in 15m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f841d9ca76934bddbdc6333fb11ae644 : FAILURE in 22m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ea839816c224dd5bb37ffec69e802f8 : FAILURE in 22m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38cad27c76344c199f0d5b0d1c12da0d : FAILURE in 22m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/09a99ac150f64ce381d620dce0a8a81c : FAILURE in 18m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bf2b2b0fb84c434e8d59a14de7d7281f : FAILURE in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1d3edf90869447489bb1632296afba03 : FAILURE in 21m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b9963d0156f458ca7fd326bb17fa635 : FAILURE in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/056006e373ba450fbddd9d5a2df66c90 : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a912e04564564ac7aa33a166f77ed63d : SUCCESS in 8m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b8c148cb634941edaff3e6ef53f1f184 : FAILURE in 58m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a4e748a02f2410ca8acba47731e67bb : FAILURE in 1h 13m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46421dcebb2b443497d85563997cb713 : FAILURE in 50m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95e48d84bab94e29804e2f7edd8083e1 : FAILURE in 56m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed3a549ae63845599c1cf7919c8da664 : FAILURE in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe99cb8061dd4054b649bc379d9fc01b : FAILURE in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/08853411887b43939e9f92f8106e44f7 : FAILURE in 23m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d3ff0c96b294593ac997191192bfd20 : SUCCESS in 39m 57s\n\nWarning:\n  Comments left for invalid file swift/common/middleware/s3api/s3api.py","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"29616b32fce1ba06c01218106347a2bd966e84d1","revisions":{"82af692ee97ffa05ca8b6d43f0e685c5c6b22b96":{"kind":"REWORK","_number":1,"created":"2021-11-01 05:10:40.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/1"}}},"commit":{"parents":[{"commit":"ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea","subject":"trace: Add RequestTraceMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-01 05:02:33.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-01 05:10:04.000000000","tz":660},"subject":"Tracing: Remove all but the OpenTracing code","message":"Tracing: Remove all but the OpenTracing code\n\nNow that we have opentracing intergrated a bunch of the original tracing\ncode, like saving to json, isn\u0027t really required anymore. So this patch\nremoves most the extra code.\n\nThe trace_data object remains as we use the start_time for delta\ncalculations, enter and leave seem interesting to have logged in tacing\nso we get some harder numbers. And the callstack in the trace_data needs\nto be there for trace name uniqueness for when we use more then one of\nthe same middleware. (at least for now).\n\nAlso fixed some time delta assumptions that was wrong.\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82af692ee97ffa05ca8b6d43f0e685c5c6b22b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82af692ee97ffa05ca8b6d43f0e685c5c6b22b96"}]},"branch":"refs/heads/master"},"a6dca7a103c10af56ef910e73d402dd1fa30414f":{"kind":"REWORK","_number":2,"created":"2021-11-08 04:30:35.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/2"}}},"commit":{"parents":[{"commit":"ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea","subject":"trace: Add RequestTraceMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-01 05:02:33.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-08 04:30:05.000000000","tz":660},"subject":"Tracing: Remove all but the OpenTracing code","message":"Tracing: Remove all but the OpenTracing code\n\nNow that we have opentracing intergrated a bunch of the original tracing\ncode, like saving to json, isn\u0027t really required anymore. So this patch\nremoves most the extra code.\n\nThe trace_data object remains as we use the start_time for delta\ncalculations, enter and leave seem interesting to have logged in tacing\nso we get some harder numbers. And the callstack in the trace_data needs\nto be there for trace name uniqueness for when we use more then one of\nthe same middleware. (at least for now).\n\nAlso fixed some time delta assumptions that was wrong.\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6dca7a103c10af56ef910e73d402dd1fa30414f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6dca7a103c10af56ef910e73d402dd1fa30414f"}]},"branch":"refs/heads/master"},"91cf1a90d2ef46b9e99379648b13a7d95cdc91ef":{"kind":"REWORK","_number":3,"created":"2021-11-11 03:37:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/3"}}},"commit":{"parents":[{"commit":"ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea","subject":"trace: Add RequestTraceMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-01 05:02:33.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-11 03:37:42.000000000","tz":660},"subject":"Tracing: Remove all but the OpenTracing code","message":"Tracing: Remove all but the OpenTracing code\n\nNow that we have opentracing intergrated a bunch of the original tracing\ncode, like saving to json, isn\u0027t really required anymore. So this patch\nremoves most the extra code.\n\nThe trace_data object remains as we use the start_time for delta\ncalculations, enter and leave seem interesting to have logged in tacing\nso we get some harder numbers. And the callstack in the trace_data needs\nto be there for trace name uniqueness for when we use more then one of\nthe same middleware. (at least for now).\n\nAlso fixed some time delta assumptions that was wrong.\n\nThis patch also does better current span tracking when we use\nconextmanagers. A tracer looses it\u0027s active span when doing through an\neventlet pile. So better tracking means we can pull what should be the\nlatest from inside the green thread, making them trace properly\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91cf1a90d2ef46b9e99379648b13a7d95cdc91ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91cf1a90d2ef46b9e99379648b13a7d95cdc91ef"}]},"branch":"refs/heads/master"},"cfcceac10790d2ff6c534cc857e042d425be62bb":{"kind":"REWORK","_number":4,"created":"2021-12-16 03:50:13.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/4"}}},"commit":{"parents":[{"commit":"ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea","subject":"trace: Add RequestTraceMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec43e3bc3a5f22c8157a5183eb9e4d6ae10859ea"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-01 05:02:33.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-16 03:49:53.000000000","tz":660},"subject":"Tracing: Remove all but the OpenTracing code","message":"Tracing: Remove all but the OpenTracing code\n\nNow that we have opentracing intergrated a bunch of the original tracing\ncode, like saving to json, isn\u0027t really required anymore. So this patch\nremoves most the extra code.\n\nThe trace_data object remains as we use the start_time for delta\ncalculations, enter and leave seem interesting to have logged in tacing\nso we get some harder numbers. And the callstack in the trace_data needs\nto be there for trace name uniqueness for when we use more then one of\nthe same middleware. (at least for now).\n\nAlso fixed some time delta assumptions that was wrong.\n\nThis patch also does better current span tracking when we use\nconextmanagers. A tracer looses it\u0027s active span when doing through an\neventlet pile. So better tracking means we can pull what should be the\nlatest from inside the green thread, making them trace properly\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfcceac10790d2ff6c534cc857e042d425be62bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfcceac10790d2ff6c534cc857e042d425be62bb"}]},"branch":"refs/heads/master"},"2b08c7418e4e2056f0d3295117c948c4e7584851":{"kind":"REWORK","_number":5,"created":"2022-01-11 01:31:05.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/5"}}},"commit":{"parents":[{"commit":"7fbe4339121a05e868f1b4137a6da5797c7cc393","subject":"trace: Add RequestTraceMiddleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7fbe4339121a05e868f1b4137a6da5797c7cc393"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-01 05:02:33.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-11 01:30:09.000000000","tz":660},"subject":"Tracing: Remove all but the OpenTracing code","message":"Tracing: Remove all but the OpenTracing code\n\nNow that we have opentracing intergrated a bunch of the original tracing\ncode, like saving to json, isn\u0027t really required anymore. So this patch\nremoves most the extra code.\n\nThe trace_data object remains as we use the start_time for delta\ncalculations, enter and leave seem interesting to have logged in tacing\nso we get some harder numbers. And the callstack in the trace_data needs\nto be there for trace name uniqueness for when we use more then one of\nthe same middleware. (at least for now).\n\nAlso fixed some time delta assumptions that was wrong.\n\nThis patch also does better current span tracking when we use\nconextmanagers. A tracer looses it\u0027s active span when doing through an\neventlet pile. So better tracking means we can pull what should be the\nlatest from inside the green thread, making them trace properly\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b08c7418e4e2056f0d3295117c948c4e7584851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b08c7418e4e2056f0d3295117c948c4e7584851"}]},"branch":"refs/heads/master"},"233a02d957c124e588b26b56fa0ca850663124d2":{"kind":"REWORK","_number":6,"created":"2022-01-31 02:57:49.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/6"}}},"commit":{"parents":[{"commit":"20024c728ea3c741b0edce3876c0c607d268a924","subject":"Prepare tracing by adding a WSGI mixin for middlewares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20024c728ea3c741b0edce3876c0c607d268a924"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-30 07:13:01.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-31 02:43:57.000000000","tz":660},"subject":"trace: Add RequestTraceMiddleware","message":"trace: Add RequestTraceMiddleware\n\nThis middleware is used to enable request tracing. At this stage the is\nno actual HMAC To start tracing so any value can be put into the\nquery_string parameter trace_sig. When you do that req will be traced.\n\nThe middleware also introduces another option which will trace every x\nrequest. This can be handy if you want to insert the middleware into\nsomething like the internal client config.\n\nThe request will start and stop tracing as the request hits this\nmiddleware. As such, to get as much info as possible it\u0027s best to have it\nas close to the front of the pipeline as possible.\n\nThe proxy has been instrumented so when making calls to storage nodes,\nso long as the middleware is also inserted into the storage node\npipelines, the tracing context is injected into the sub requests allowing\nus to trace a single request throughout the cluster.\n\nRequest tracing utilises the OpenTracing framework. So if you want to\ntrace there is a run time dependency on python opentracing module.\n\nThere is a trace_data object which contains the start_time for delta\ncalculations, enter and leave seem interesting to have logged in case we\nwant access in the code (though I might still remove this as we get\ntimings as part of opentracing). The data structure also has a callstack\nwhich is being used for name uniqueness. This data structure was used\npre OpenTracing was implemented. And might be removed before the final\nversion, but some code currently still requires it, so haven\u0027t removed\nit... yet.\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/233a02d957c124e588b26b56fa0ca850663124d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/233a02d957c124e588b26b56fa0ca850663124d2"}]},"branch":"refs/heads/master"},"059b9d4512ab44217f00aa7833737901d11cc1bb":{"kind":"REWORK","_number":7,"created":"2022-03-09 22:37:58.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/7"}}},"commit":{"parents":[{"commit":"89370783773ed745f8475d73c31c8fb61d8110ac","subject":"Prepare tracing by adding a WSGI mixin for middlewares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89370783773ed745f8475d73c31c8fb61d8110ac"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-30 07:13:01.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-03-09 22:37:32.000000000","tz":660},"subject":"trace: Add RequestTraceMiddleware","message":"trace: Add RequestTraceMiddleware\n\nThis middleware is used to enable request tracing. At this stage the is\nno actual HMAC To start tracing so any value can be put into the\nquery_string parameter trace_sig. When you do that req will be traced.\n\nThe middleware also introduces another option which will trace every x\nrequest. This can be handy if you want to insert the middleware into\nsomething like the internal client config.\n\nThe request will start and stop tracing as the request hits this\nmiddleware. As such, to get as much info as possible it\u0027s best to have it\nas close to the front of the pipeline as possible.\n\nThe proxy has been instrumented so when making calls to storage nodes,\nso long as the middleware is also inserted into the storage node\npipelines, the tracing context is injected into the sub requests allowing\nus to trace a single request throughout the cluster.\n\nRequest tracing utilises the OpenTracing framework. So if you want to\ntrace there is a run time dependency on python opentracing module.\n\nThere is a trace_data object which contains the start_time for delta\ncalculations, enter and leave seem interesting to have logged in case we\nwant access in the code (though I might still remove this as we get\ntimings as part of opentracing). The data structure also has a callstack\nwhich is being used for name uniqueness. This data structure was used\npre OpenTracing was implemented. And might be removed before the final\nversion, but some code currently still requires it, so haven\u0027t removed\nit... yet.\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/059b9d4512ab44217f00aa7833737901d11cc1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/059b9d4512ab44217f00aa7833737901d11cc1bb"}]},"branch":"refs/heads/master"},"29616b32fce1ba06c01218106347a2bd966e84d1":{"kind":"REWORK","_number":8,"created":"2022-09-12 00:31:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/19/816119/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/816119/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/816119/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/816119/8"}}},"commit":{"parents":[{"commit":"0f3a9ed2f86309e0236de2040fef734366441001","subject":"Prepare tracing by adding a WSGI mixin for middlewares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f3a9ed2f86309e0236de2040fef734366441001"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-07-30 07:13:01.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-09-12 00:29:15.000000000","tz":600},"subject":"trace: Add RequestTraceMiddleware","message":"trace: Add RequestTraceMiddleware\n\nThis middleware is used to enable request tracing. At this stage the is\nno actual HMAC To start tracing so any value can be put into the\nquery_string parameter trace_sig. When you do that req will be traced.\n\nThe middleware also introduces another option which will trace every x\nrequest. This can be handy if you want to insert the middleware into\nsomething like the internal client config.\n\nThe request will start and stop tracing as the request hits this\nmiddleware. As such, to get as much info as possible it\u0027s best to have it\nas close to the front of the pipeline as possible.\n\nThe proxy has been instrumented so when making calls to storage nodes,\nso long as the middleware is also inserted into the storage node\npipelines, the tracing context is injected into the sub requests allowing\nus to trace a single request throughout the cluster.\n\nRequest tracing utilises the OpenTracing framework. So if you want to\ntrace there is a run time dependency on python opentracing module.\n\nThere is a trace_data object which contains the start_time for delta\ncalculations, enter and leave seem interesting to have logged in case we\nwant access in the code (though I might still remove this as we get\ntimings as part of opentracing). The data structure also has a callstack\nwhich is being used for name uniqueness. This data structure was used\npre OpenTracing was implemented. And might be removed before the final\nversion, but some code currently still requires it, so haven\u0027t removed\nit... yet.\n\nChange-Id: Ibb2cfa6e324de0801468860ab9a71edb94ef5c99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29616b32fce1ba06c01218106347a2bd966e84d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29616b32fce1ba06c01218106347a2bd966e84d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
