)]}'
{"id":"openstack%2Fswift~817294","triplet_id":"openstack%2Fswift~master~I0e8c0655f09af4a201c905098b2ead970aa52568","project":"openstack/swift","branch":"master","topic":"p-cache-shards","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-12-15 21:17:28.000000000","reason":"Alistair Coles replied on the change"}},"hashtags":[],"change_id":"I0e8c0655f09af4a201c905098b2ead970aa52568","subject":"container-server: set shard ranges in memcache","status":"NEW","created":"2021-11-09 18:24:35.000000000","updated":"2021-12-15 21:17:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"b45811bc803eba331632d330fcf6cbf4c81ffb34","_number":817294,"virtual_id_number":817294,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-29 18:07:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2021-12-15 21:17:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-09 19:37:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 19:45:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"b87e0d88d8186a3d81d28b17e94f0fdae51e03e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-09 18:24:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b0da36a2e701fc4a295a3d40610b6f8b05fea85","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-09 18:25:16.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52eb68c1a1ef4a1637a71907eb528ce1c40743af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:37:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df0c78d158448e0aad1a04bcb945b1e : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/6b44ca8b2de04f3b9299ba7034ebed94 : SUCCESS in 1h 06m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f70fa86502c44df6be4e9b64dc9d257a : SUCCESS in 40m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8f184952a4c4fc1b84ca85d9817587b : SUCCESS in 11m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9d6ea6a0f58c4aa182b9ca096f2edd8f : SUCCESS in 7m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fed3b682d96441dba339ed74f5999130 : SUCCESS in 18m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0a111a0c166473c92347333b41e2ff7 : SUCCESS in 15m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e3a6b476379e4c8c987aa731cd7473a2 : SUCCESS in 15m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/316561fe090448a4a2b757855f518872 : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3d41da295ee41f0814dcc67804d6c77 : SUCCESS in 13m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4dd44a770cb2467f8e0bb40b5d06a999 : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ba403fa6c914851b6d2be7516b7d222 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d4d238c945e4c079e3feb6d5c64ef02 : SUCCESS in 17m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/635f982f062e418b81cd08c0cfce0cf5 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f22074dfbeb64d688b7f597241596400 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cd1b171b123c4e198feaa727f9e85f9c : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/91fd89d96f98404981eb2641097f44c4 : SUCCESS in 8m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/391280996eaf4e038c6e401386fd7e2d : SUCCESS in 23m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6fec13a8edb0442fbca70380157765a7 : SUCCESS in 49m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b4cc3e32c0494c6bbbbd266051a2e316 : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7800e7acc7424816bbc3de0b36a575f1 : SUCCESS in 38m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1060671aefb5475ea354416920012bf9 : SUCCESS in 44m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62c299b990f349e9859d29c5bb5b123c : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6459eef4ff4d4112bda7b507e981d22c : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8707d48063564d2e81931df9cdcd1b0c : SUCCESS in 22m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79772602fc7b48b98d03c7d9bf88cb21 : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":1},{"id":"512328061ae12c7208af8f93f03318eba7168b12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 20:06:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83e836cf7c894ddbbdbd2d6d1c94831d : SUCCESS in 29m 00s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7aa6a33bd9704288889538e14a23b9da : SUCCESS in 32m 11s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/c513d59cfff9486dafa5a5e92bc5e5fe : FAILURE in 1h 36m 13s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5993ef2808664421b3555a7af6e72047 : SUCCESS in 56m 09s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/86f06e77f7e24be5a9c8cc202fa30e08 : SUCCESS in 32m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1125126b39b234884b6d9aaa4a1a982cf4757d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-29 16:54:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c0ced923f56f84850c8d602db4e40a17cf9b49b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 18:07:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a347aae3611f49fdb371568236ba4ab1 : SUCCESS in 10m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/7026da7cd1c543bdb1036532a470946a : SUCCESS in 59m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0bfe61acfa644fee9f07b736f59576c3 : SUCCESS in 38m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7234bb4bd6224c74871dbcb8f607925b : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8fe0f1062c1f43ae8ce172b87ebc702f : SUCCESS in 8m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fab6f04eabd5499d83afa6ee0b220c88 : SUCCESS in 15m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1c2c4abdfba40c1801efac01f0cdb42 : SUCCESS in 16m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c22f5341fb664466941bd84b35ac934a : SUCCESS in 15m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c97219fb345c4d1c882eb0641b5d2a59 : SUCCESS in 15m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37385a00b4be4d1f9e64ee64eb4df272 : SUCCESS in 14m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33f976e7432b44cd885ba63bdf91d083 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b049723e05946e1bd1cd16d143725c3 : SUCCESS in 15m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10e2b8b848154d3e912525e1d1c879f8 : SUCCESS in 21m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5a2db253af204a2dbaaa4b3a553dfa4a : SUCCESS in 21m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/250f35c4a65c44149a5dc55503ac20ef : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1157bc0fa34d44cc8dfd8212090d6301 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c34b22cfc0f040d2b44bf716325c90af : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af5744811cf54e6aa3863cfc22df153c : SUCCESS in 19m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d0164d08627c44edb17683e420746013 : SUCCESS in 59m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f7e8ae4516d74cbd8865171eb7940fdf : SUCCESS in 1h 04m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fcbfccf6646a4505957b3c26204fcc9a : SUCCESS in 40m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11e9ef5290aa4244b02a56166386c146 : SUCCESS in 32m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4c15732d05f4211930c61dbd85217ee : SUCCESS in 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eab1b5ca049346a4be7a8d5a00921e8b : SUCCESS in 8m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/43547a2320a34b2e8fdc7ee3f483262a : SUCCESS in 23m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae8a62d67479486391a1ee40e105c9d8 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":2},{"id":"7c51158f75beabd0def2ad41da2378a4c5f1ce36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 20:06:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fd109b1fb0f439297e234973d5bf103 : SUCCESS in 29m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f3433dc5328445a830ccf82bae3c8f7 : SUCCESS in 1h 03m 55s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/829d280c74bc4f078e123c1d5994b80b : SUCCESS in 1h 28m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6599a37daadf41c388617f6d90394802 : SUCCESS in 1h 07m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/87199db214864f62bdf43df01303c525 : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":2},{"id":"aa718ff43027b14210ecd0d90f3d7c17b65a9f3c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-08 19:45:25.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b45811bc803eba331632d330fcf6cbf4c81ffb34","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-15 21:17:28.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b20475bb06483d7863b1bf3d25c15fe0574cbbe","revisions":{"03aec6b372f60081e25c6903c3567bb626a62943":{"kind":"REWORK","_number":1,"created":"2021-11-09 18:24:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/817294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/817294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/817294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/817294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/817294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/817294/1"}}},"commit":{"parents":[{"commit":"27478f07c3f45067a2644140b7da2b356ca9ac60","subject":"Merge \"proxy: Remove a bunch of logging translations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27478f07c3f45067a2644140b7da2b356ca9ac60"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-09 15:58:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-09 18:24:13.000000000","tz":0},"subject":"WIP container-server: set shard ranges in memcache","message":"WIP container-server: set shard ranges in memcache\n\nThe proxy-server puts shard ranges into memcache to avoid making a GET\nrequest to container servers on every object PUT for the purposes of\ndirecting object updates to the correct shard. On the occasions when\nthe memcache entry has expired and the proxy does make a GET request\nto the container server, a container server under extreme load may not\nrespond in a timely way. A primary container servers may eventually\nrespond to the proxy GET requests, but the proxy will have timed\nout, gone to handoffs and given up on the container GET.\n\nThis has two undesirable consequences: the proxy cannot repopulate\nmemcache with shard ranges, and object updates default to being\ndirected to the root container, therefore placing more load on the\nalready busy container servers.\n\nThis patch enables the container server to set shard ranges into\nmemcache if it has access to the memcache pool. That way, even when\ncontainer server responses are slow and the proxy times out, the shard\nranges may still make it to memcache for subsequent proxy requests to\nuse.\n\nChange-Id: I0e8c0655f09af4a201c905098b2ead970aa52568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03aec6b372f60081e25c6903c3567bb626a62943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03aec6b372f60081e25c6903c3567bb626a62943"}]},"branch":"refs/heads/master"},"5b20475bb06483d7863b1bf3d25c15fe0574cbbe":{"kind":"REWORK","_number":2,"created":"2021-11-29 16:54:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/94/817294/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/817294/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/817294/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/817294/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/817294/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/817294/2"}}},"commit":{"parents":[{"commit":"8bda234e40a28ab5b526ae4f62b0e26c5ba66e29","subject":"Merge \"Make arm jobs voting (but not the pipeline)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bda234e40a28ab5b526ae4f62b0e26c5ba66e29"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-09 15:58:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-29 16:54:06.000000000","tz":0},"subject":"container-server: set shard ranges in memcache","message":"container-server: set shard ranges in memcache\n\nThe proxy-server puts shard ranges into memcache to avoid making a GET\nrequest to container servers on every object PUT for the purposes of\ndirecting object updates to the correct shard. On the occasions when\nthe memcache entry has expired and the proxy does make a GET request\nto the container server, a container server under extreme load may not\nrespond in a timely way. A primary container server may eventually\nrespond to the proxy GET requests, but the proxy will have timed out,\ngone to handoffs and given up on the primary container server GET.\n\nThis has two undesirable consequences: the proxy cannot repopulate\nmemcache with shard ranges; object updates default to being\ndirected to the root container, therefore placing more load on the\nalready busy container servers.\n\nThis patch enables the container server to set shard ranges into\nmemcache if it has access to the memcache pool. That way, even when\ncontainer server responses are slow and the proxy times out, the shard\nranges may still make it to memcache for subsequent proxy requests to\nuse.\n\nChange-Id: I0e8c0655f09af4a201c905098b2ead970aa52568\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b20475bb06483d7863b1bf3d25c15fe0574cbbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b20475bb06483d7863b1bf3d25c15fe0574cbbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
