)]}'
{"id":"openstack%2Fswift~817345","triplet_id":"openstack%2Fswift~master~I66f6fd712cd14f60be39e2ee7ddd0a089cedabab","project":"openstack/swift","branch":"master","topic":"cleanup-devstack-jobs","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2021-11-10 09:43:01.000000000","reason":"Removed by Dr. Jens Harbott (frickler) by clicking the attention icon"}},"hashtags":[],"change_id":"I66f6fd712cd14f60be39e2ee7ddd0a089cedabab","subject":"Clean up devstack job definitions","status":"MERGED","created":"2021-11-10 05:40:04.000000000","updated":"2021-11-10 19:51:25.000000000","submitted":"2021-11-10 19:41:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817345-cleanup-devstack-jobs","meta_rev_id":"19be41c9db14631855ac3b91b48787acfdf6f858","_number":817345,"virtual_id_number":817345,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-10 19:41:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-10 05:40:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-10 05:40:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-10 05:40:58.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-11-10 06:48:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ab8365794859a02931afb009c9b6242ef58b7f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-10 05:40:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c29e57cf450e396c894fac0f7c2ef61fff02960","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-10 05:40:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb74b7733591b825028fdaf842e340de4e8d68f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 06:48:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b01576775a43abac7e54ac7c3b2d69 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/a994d4ebd9fd4dd69b5a009c17276036 : SUCCESS in 1h 00m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/263693fd7a3c4a9a9854542b18aee80d : SUCCESS in 51m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca7ad9324c3b4afc91872aaf2f29705b : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bcef5b5ae2eb4762aafeb87fe527380a : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ebf0a19e7a8f4ffcb72f0e55359d7c9e : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/598b8f6bdd47487dbd7f29338bcee574 : SUCCESS in 17m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/665204b9cc454cd8b40f474d35b3d4db : SUCCESS in 17m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bdef16716786443c804dc0b84941b20a : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55b658208bfe48b5b4f6afc85a1cb3e4 : SUCCESS in 15m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be07d3191b34432ba795b1265b525d0e : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0809721bb8444f38a61671b2e27e7ced : SUCCESS in 16m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80273e77e43748e78be88bf8fbbc09d2 : SUCCESS in 19m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d55648ab823e498d970c420bd2ee65d5 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77bbd23e308a4835923523d3e6d20c2f : SUCCESS in 14m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a6e16b0da8924bea9c9d107858889136 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8dc8aded2ebb42ceb4d1080378c78703 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99d33fec08e64ed89f1c10aeb212940c : SUCCESS in 20m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4df9fe7ab6fb4fa1981469d241af0bdc : SUCCESS in 48m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c48b9c507ce84a6f9646a986ac9c07e3 : SUCCESS in 1h 02m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c624a737d67c4cd59e5cbcc6cda11d4a : SUCCESS in 52m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a2982c50a21a4e6ea92c7f7c925417da : SUCCESS in 50m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bce5bfe0edfc4ccc91890f67570ce981 : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628d18c8d14f439fac77f36aa4fcaaf6 : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a011775465784f2b91ca1daaf67bb7d6 : SUCCESS in 23m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6242150b8c6a4c3095d1236425fde41c : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e549cb3ce48413ca059ad25e2eb1bb309ef6c104","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 18:25:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d58ba9679ed454f8ff5eaee3e4a4ff9 : SUCCESS in 50m 31s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fad3c111dce64d0aa1f4fe7f4ac03257 : SUCCESS in 34m 13s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/3dfd53111a664ecbabed87beb1e0b057 : FAILURE in 2h 18m 38s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbad7f050f8144f48cdbbe21b6c35df1 : SUCCESS in 33m 02s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e26651dcaa0494bb7f73c00522b6077 : SUCCESS in 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd60cdeab6ec149b39bed957c76e61d218d0ace7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 18:26:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dda076a044323374a25e7452c0a102f9eafd1003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 19:41:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16030b2a9ebe411383c11a29500bbfb0 : SUCCESS in 11m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/0e9048fb23f84920a9648dbc13675681 : SUCCESS in 1h 05m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9b9448c0c224990b7f6122e658d9330 : SUCCESS in 48m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/39a1a257207e439e992f3e681bee008d : SUCCESS in 11m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/0bc0f6d3e4614f5ebf5f130f07356066 : SUCCESS in 13m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0203b928f2e470194ec3497fedf0547 : SUCCESS in 32m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/32a844dad16c464c8eb53c251da47a4b : SUCCESS in 19m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8a054209f9394e1a9ff3f8e4d48abb82 : SUCCESS in 22m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/34646a284ec54d339f5f114f1b2d3e31 : SUCCESS in 20m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1034a32470c24920abf9ac38f755709c : SUCCESS in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/670c0f26c1c94f64b04a0137f904c4fe : SUCCESS in 21m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9fed91333622410caefc9f6fdea4e894 : SUCCESS in 25m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/93e5839f557743ca86347b34323bf140 : SUCCESS in 22m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b862c3ddde3f4b87a11a2ab27e89a393 : SUCCESS in 34m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f068b8bf6a3544a090b18f830b750e50 : SUCCESS in 24m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cdda1e41d1e945218478fed826cb7157 : SUCCESS in 23m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c522fdaf9994a889b7a3e78b85df38b : SUCCESS in 7m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d541471f9ac14e52865dd0e356ef3f06 : SUCCESS in 51m 40s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/91c16f4f2f3a4d09922aa3d0a5cb5cd7 : SUCCESS in 1h 07m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e7abb36ed7d2427c866e3d6ab3a2a4a8 : SUCCESS in 37m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7bc4f5913be1477f8e264ca790a2eaf4 : SUCCESS in 51m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e1e65a6c2684f6aa4d73871a526e806 : SUCCESS in 19m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864129d209314d6c92be6aa375f7f63b : SUCCESS in 7m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d0914818c1447bd9a4e71d6940c1eb3 : SUCCESS in 25m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb9c77b1f6644c749091c3aeedd4cd6b : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7e5020dd6d8f2dbc86e2fc5e38af25e7ca64a5c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 19:41:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"19be41c9db14631855ac3b91b48787acfdf6f858","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 19:51:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f500451f9d6496db4b9264a29296f67 : SUCCESS in 1m 37s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ac0e5aec993041a48f4c06781fac4afa : SUCCESS in 1m 05s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d76c99112d874389aa0eee8868bdcae4 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d190dba5d3889c4b9950e074335e5f5bb137437","revisions":{"3d190dba5d3889c4b9950e074335e5f5bb137437":{"kind":"REWORK","_number":1,"created":"2021-11-10 05:40:04.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/45/817345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/45/817345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/817345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/817345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/45/817345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/45/817345/1"}}},"commit":{"parents":[{"commit":"27478f07c3f45067a2644140b7da2b356ca9ac60","subject":"Merge \"proxy: Remove a bunch of logging translations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27478f07c3f45067a2644140b7da2b356ca9ac60"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-10 05:38:21.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2021-11-10 05:39:52.000000000","tz":60},"subject":"Clean up devstack job definitions","message":"Clean up devstack job definitions\n\nKeystone can only provide v3 now, remove variables that no longer have\nany effect within devstack.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I66f6fd712cd14f60be39e2ee7ddd0a089cedabab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d190dba5d3889c4b9950e074335e5f5bb137437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d190dba5d3889c4b9950e074335e5f5bb137437"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
