)]}'
{"id":"openstack%2Fswift~817710","triplet_id":"openstack%2Fswift~master~Icdcd429b662bdea59cb0d418b7e026ab9b896d1e","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"32291":{"account":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"last_update":"2021-12-22 01:05:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icdcd429b662bdea59cb0d418b7e026ab9b896d1e","subject":"Replace deprecated assertRegexpMatches","status":"ABANDONED","created":"2021-11-12 07:43:06.000000000","updated":"2021-12-22 01:05:43.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"72a51fd3166712fee5369c7ae19921616607806b","_number":817710,"virtual_id_number":817710,"owner":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-12 08:25:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-12 08:25:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd98b6a70f185c41b8200c0a1b08c179db7999a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-11-12 07:43:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3ecc0f58cdd03972ad7a8d5265d72d6b54fdc30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 08:25:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09806cca5c3c4902840b82bd5ec165b3 : SUCCESS in 10m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e0e9c0015974f8490a3490f8a764890 : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/af74f732d504403a89e0215d01b6e971 : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12d62e8083084958830d4e81d4455520 : FAILURE in 28m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/092b9b51be02468f8af69400d486016d : SUCCESS in 29m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eebfc47aa3fc4f0db082a9b37f470426 : SUCCESS in 27m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7263bafb4b5240b9a7a41a9e1ae743f6 : SUCCESS in 26m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac8193cb9d9b453bbb63a19187f9b548 : SUCCESS in 26m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ed7a670391844a18a4bbf57de22f685 : FAILURE in 20m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5b9c60719454af4af0d9f44e3c254db : FAILURE in 14m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f6b1dbc876644f369fffb6275e9e8354 : FAILURE in 17m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/224b407c378243e09e0283aaf8faa34c : SUCCESS in 39m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bd2c51c10a0e470f8d8a4baa3643fba8 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1897b58f60e84605a0fbb334967ae4bd : SUCCESS in 14m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd7a52219ff14e18b564a2a9be63c67f : SUCCESS in 20m 46s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c41757e79044f79b3b2cda8955e6e67 : SUCCESS in 36m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/29fadb51e965415d8758944aa669aeab : SUCCESS in 32m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ab99288439f43dcb6288d2b0e16af0e : FAILURE in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa6df36bf6d4e5fb1cd12c73fbe22bb : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35134177eed24e059e36b8d959010762 : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7ab52563c95a7ae8de62071f3812f4a1f35f0870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 09:47:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f19aff7639f4126ad73e3a91794fdb2 : SUCCESS in 37m 49s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8fa04c36fe847d98d490ff2347d0408 : SUCCESS in 46m 26s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/8502b40c70244e4aa39e5907f320687e : FAILURE in 2h 01m 46s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3397fcc6406d4c83a9820ae3539fde85 : SUCCESS in 38m 14s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f489d0074e8443ea2d7a103f1a986f6 : SUCCESS in 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef110ec347c4a7204684f11e7d9af54e2c90a770","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-11-15 01:06:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10aee6d65a00f98b175322a45626bb61d852093f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 01:55:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62cdc8a93c2483c8ae1f320cf236aef : SUCCESS in 9m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c669f127ec7649e2b9a5c4b90967897c : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/92b102ccdda94b9c9deedac369a302df : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c341189e65274aeb8896ef3e00a005b7 : FAILURE in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/754e90db4c3f471680ea3234ad2ca169 : SUCCESS in 15m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c6313f316e98426a8823287dd0b6b8ec : SUCCESS in 15m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/54cc2e78ee4143ca8b69a49d0b14bf84 : SUCCESS in 18m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac19eccd0f4844739cb1fee826f4cce4 : SUCCESS in 16m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2edc31784e814188bfd2c2cc4f6fc591 : FAILURE in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e99a1b87c8094c818030da6f47921143 : FAILURE in 20m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac11bc5c374e48f89f15c632aea944f9 : FAILURE in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3610a4dbdac948a5ac78ccef2eb6e5d8 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8701506b655420cb6022d8bd44f92ed : SUCCESS in 21m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1c3d4b8a67a44959aecb31d20f4bab1f : SUCCESS in 16m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/33c70679aefb41598803642216cff0ba : SUCCESS in 18m 49s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/357cd1a2c60e400bba7a20e1120f0695 : SUCCESS in 43m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1620fcce77e24569adfd3af8584f725e : SUCCESS in 40m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63d485c21cc2452c9d8860134a822401 : FAILURE in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9a7add50ac4a0c82e195420093dc6b : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/37c4319ed2cc46479ef4b06dc8a372af : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c597969b69e9be83d67743b1e2c6cb61fcd8ec9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 02:54:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b914f36ecdff431da90243a12b88f851 : SUCCESS in 46m 43s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b46ac81a9bfb4f39a5422b16cc5751c1 : SUCCESS in 41m 19s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/7f9f10e0bf48475ca229dca105cddb0e : FAILURE in 1h 39m 46s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/934554848b744e1db0efba69ec04fbe2 : SUCCESS in 48m 46s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d252a801a97f4c339b6fc5e2ca986316 : SUCCESS in 42m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26f1fdbd02fd1cc931f137cbca80f831fafd9e99","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-21 07:14:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e140238a0263e687e56af9d504110f85f46d2e89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 08:06:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74590c3b6c54b1e8e67381ce7713cac : SUCCESS in 7m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79fdf562bf2f4efdbabea4d431ebf833 : SUCCESS in 38m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/10b6b9c9add04fb7b8a175e069a03e31 : SUCCESS in 8m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/96bb1a5aa960485fbf99a692d9fe514a : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97a1c6eb107a45798c3a70a8efac518d : FAILURE in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c7de555ccb7f4c8389663866ad877ef7 : SUCCESS in 15m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9ad3c40e1f91479f81c719021d1019eb : SUCCESS in 12m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1200401cb84d4b17a9954c4cfc47c936 : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4b0389a86b344eee99ec9e243c65e5e0 : SUCCESS in 17m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d85271535e7443fae26d7076ae07af4 : FAILURE in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/95199a1c02034ae7be5b05c4e3ee3b45 : FAILURE in 19m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2c2b2de01ab8423091b9624ea81d16bf : FAILURE in 19m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b8b7f3a6c0cd4591b4f5ead988c81e03 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5e8d241ad24a4dabb971794e3464a9ba : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/08009bb1375a44fe85dc03939c22cae3 : SUCCESS in 17m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c176ee12a9ce413babe5de96c22ad5d8 : SUCCESS in 37m 10s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62bd43e6f48e46718c516cace9406e1d : SUCCESS in 46m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c6c707e98114539bf61c2fe17a01cc6 : SUCCESS in 38m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a525a41f06a4141897d3fdc62935795 : FAILURE in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8a47b2b923b4994bbd376bdbbef65d4 : SUCCESS in 7m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f522404ef3cd4dbfac77b0c9a1e78948 : SUCCESS in 20m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6533b5418ee168c573c0ebb2fda13cdd2f47b74b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 09:36:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f58352a414c4241b6e6ebb8b868208e : SUCCESS in 48m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49cc4330e1a641678f3f554dd75a4eec : SUCCESS in 39m 25s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c5ed81763a9e405093ab6b0a1989ff42 : FAILURE in 2h 18m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae5c9f9737424b8ea5a257ff1e444e71 : SUCCESS in 50m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb3a6420e6fd481aaf55040f97aa3a3b : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"72a51fd3166712fee5369c7ae19921616607806b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-12-22 01:05:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c065dbfe711ec4aecc308650f748e49c8db8d0b","revisions":{"6c065dbfe711ec4aecc308650f748e49c8db8d0b":{"kind":"REWORK","_number":1,"created":"2021-11-12 07:43:06.000000000","uploader":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"ref":"refs/changes/10/817710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/817710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/817710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/817710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/817710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/817710/1"}}},"commit":{"parents":[{"commit":"3d190dba5d3889c4b9950e074335e5f5bb137437","subject":"Clean up devstack job definitions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d190dba5d3889c4b9950e074335e5f5bb137437"}]}],"author":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-11-12 05:56:15.000000000","tz":480},"committer":{"name":"wu.shiming","email":"wushiming@yovole.com","date":"2021-11-12 07:42:33.000000000","tz":480},"subject":"Replace deprecated assertRegexpMatches","message":"Replace deprecated assertRegexpMatches\n\nThe assertRaisesRegexp method has been deprecated since it was renamed\nto assertRaisesRegex in Python 3.2.\n\nhttps://docs.python.org/3/library/unittest.html#deprecated-aliases\n\nChange-Id: Icdcd429b662bdea59cb0d418b7e026ab9b896d1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c065dbfe711ec4aecc308650f748e49c8db8d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c065dbfe711ec4aecc308650f748e49c8db8d0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
