)]}'
{"id":"openstack%2Fswift~818025","triplet_id":"openstack%2Fswift~master~Ia49ee158736f6ad0b73797680242b06165abd9d1","project":"openstack/swift","branch":"master","topic":"coop-reconstructor","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2021-11-16 16:06:48.000000000","reason":"Alistair Coles replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia49ee158736f6ad0b73797680242b06165abd9d1","subject":"ssync: don\u0027t break chunked protocol on error","status":"NEW","created":"2021-11-15 22:26:29.000000000","updated":"2021-11-16 16:06:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"66855e8d97a68ae7d32424fee8b00af31d5986da","_number":818025,"virtual_id_number":818025,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-16 00:04:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-11-16 16:06:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2021-11-16 15:57:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 00:04:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-16 16:06:48.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"29c0c617b78bef9c7eedbb64af7b685b20a71d01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-15 22:26:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bf9df0c756853065fe808a4a1d0bff415b353d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-15 22:59:23.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cef7a05d6047ad14f4e287ca2e4395c298d586f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 00:04:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633b2852d0bc406c8ecf62da74529ba1 : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/f447f34d7e68408995396945a01ae1c8 : SUCCESS in 1h 04m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4324c9dbd8f54334950bc4d7498d469d : SUCCESS in 44m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c030df9b263040eb87c487b3db4fab4a : SUCCESS in 8m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4d3cccda1c974f84a9f81da9a6d5f575 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d3d319d48684c12bd8991038d625e07 : FAILURE in 5m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/338288c23e6346678d598f4342668ac2 : SUCCESS in 16m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/580e7abc645c45dd9d997102cbde864a : SUCCESS in 18m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/580b74f5b72e49d884a5427328d65dba : SUCCESS in 15m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ef9633b68d242679f86bd60aa55e03f : SUCCESS in 15m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a226171d66a48ee871f6fd796d90292 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/de3ca544ecf34e62aeaeaa9a65ea88e1 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd328db4e2eb450080f27bce4ee68022 : FAILURE in 4m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/95c9c02b65a74ec49b17f3cf3c8ee3cd : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ff96247268f14c02b4833855b9071c7b : SUCCESS in 17m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/32afe6d8116f40ae89a6e5069049b192 : SUCCESS in 16m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc4345499b3f4d50869878059f85b5fe : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d1314aef0fd74640be9d396ff0a9c8df : SUCCESS in 24m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/454401b84a294ddaabec84606ec04dba : SUCCESS in 43m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/693fcc38028d49878acabf778de9da91 : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1deb6e7a9bc24b02adf0fd45aaba616a : SUCCESS in 40m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/238aadab918d4815a90eb8ef3c638011 : SUCCESS in 39m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/893f443132704076a957cd51c8366b55 : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef761810ebc842bcbb7b1f1c2e83b46f : SUCCESS in 7m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6cf74312e9b540a5a7bfdb871de7d1d7 : FAILURE in 15m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/753b8e68409d4692a331a533250f4814 : FAILURE in 19m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d157d3b08fd3c9d9e29ba4f757deca21f413001f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 00:36:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbd2511b49f34961b28befd0d97c7ac7 : SUCCESS in 28m 08s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/013659f789924e1b8c880b9a6b698bf4 : SUCCESS in 33m 49s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/e4410ecedc304ed8babeb03ee7910f23 : SUCCESS in 1h 29m 09s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f16800e4ebcb45b88bf7fc60d05d6bbf : SUCCESS in 34m 02s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a283f7071664b11bb5db4c6096eaedd : SUCCESS in 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ccbf0643c326393987e4bc0ac313c95ce4c572be","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-16 15:57:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"66855e8d97a68ae7d32424fee8b00af31d5986da","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-16 16:06:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e823151b7c7bb3a4904278d4a6b2ed302d18d62","revisions":{"837cfd472c792556fc4e8703587459c1a6eefa98":{"kind":"REWORK","_number":1,"created":"2021-11-15 22:26:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/25/818025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/818025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/818025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/818025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/818025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/818025/1"}}},"commit":{"parents":[{"commit":"591adc845909e97fd3fbcb982262eaec0ecf89ac","subject":"Fix SSYNC update phase blocking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/591adc845909e97fd3fbcb982262eaec0ecf89ac"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-11-15 22:09:42.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-11-15 22:09:42.000000000","tz":-360},"subject":"ssync: don\u0027t break chunked protocol on error","message":"ssync: don\u0027t break chunked protocol on error\n\nOnly send the last chunk message when the connection is being shutdown\ngracefully.  If we send the final chunk message when shutting down a\nconnection in the middle of another (failed/timedout) chunk we break\nprotocol and confuse the reciever leading to confusing tracebacks.\n\nChange-Id: Ia49ee158736f6ad0b73797680242b06165abd9d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/837cfd472c792556fc4e8703587459c1a6eefa98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/837cfd472c792556fc4e8703587459c1a6eefa98"}]},"branch":"refs/heads/master"},"6e823151b7c7bb3a4904278d4a6b2ed302d18d62":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-15 22:59:23.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/25/818025/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/818025/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/818025/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/818025/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/818025/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/818025/2"}}},"commit":{"parents":[{"commit":"591adc845909e97fd3fbcb982262eaec0ecf89ac","subject":"Fix SSYNC update phase blocking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/591adc845909e97fd3fbcb982262eaec0ecf89ac"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-11-15 22:09:42.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-11-15 22:58:52.000000000","tz":-360},"subject":"ssync: don\u0027t break chunked protocol on error","message":"ssync: don\u0027t break chunked protocol on error\n\nOnly send the last chunk message when the connection is being shutdown\ngracefully.  If we send the final chunk message when shutting down a\nconnection in the middle of another (failed/timedout) chunk we break\nprotocol and confuse the reciever leading to confusing tracebacks.\n\nPartial-Bug: #1621390\n\nChange-Id: Ia49ee158736f6ad0b73797680242b06165abd9d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e823151b7c7bb3a4904278d4a6b2ed302d18d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6e823151b7c7bb3a4904278d4a6b2ed302d18d62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
