)]}'
{"id":"openstack%2Fswift~818127","triplet_id":"openstack%2Fswift~master~I84f59256df3aa41338df9829a565ae830ee8e847","project":"openstack/swift","branch":"master","topic":"p-arm-jobs-voting","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-11-26 00:05:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I84f59256df3aa41338df9829a565ae830ee8e847","subject":"Make arm jobs voting (but not the pipeline)","status":"MERGED","created":"2021-11-16 17:46:05.000000000","updated":"2021-11-26 00:09:08.000000000","submitted":"2021-11-26 00:05:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"818127-p-arm-jobs-voting","meta_rev_id":"b7a56753e4ece3211e0e16ecfa8392743b4bc10e","_number":818127,"virtual_id_number":818127,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-26 00:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-25 23:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-25 23:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 19:08:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 21:17:22.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-11-25 23:08:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"b5f6185f4de7c650376a5825ee34d5fb22df7914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-16 17:46:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d332a56b68f4c4c860beb5987968b06ad18c36b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-16 17:47:11.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"683f7859c2d7419f0015ce58115a1a7af43909bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 19:08:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bca773c44ba045ac81bf3a91f612d01f : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0d121e57f4124ec292175ae7e7e1d351 : SUCCESS in 55m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05cfe7bf14144b0fb998413a3ffe16c7 : SUCCESS in 46m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b523bbbc59a41029299d3ef677a6345 : SUCCESS in 11m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/821d4f5883af4fa6a1f68c16f7b54649 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78df386a67d047929056d2ab6511cdaf : SUCCESS in 16m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ca7f50a455c49db9d6b4e14c662fad8 : SUCCESS in 15m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e09c426eb3a3456f9cc82594b590f7c1 : FAILURE in 5m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/40ab597061c14b279f998f643479d707 : SUCCESS in 15m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0bae2bd244da435f9ebd2305320b7336 : FAILURE in 5m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1b8370d6c442491094c7da4242c478fd : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/406f0aa615a14d7da0d52637ab8ea83c : SUCCESS in 21m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a455f6eedb324b339fac62addc5d44d6 : SUCCESS in 18m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c2566eec72ac4e6bae3951e8648d5246 : SUCCESS in 21m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c9a7fe81c6d24b928bf0b6170d40e0d0 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c79f5a15322842dca95f0f635319e023 : FAILURE in 4m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a612e7db4bf84dda875700cba5c0a78f : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7e23433edd04ec3ad23a4d24af2b813 : FAILURE in 6m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/298873f4a19e4e0cbc26a397175c0150 : SUCCESS in 49m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b6ca9f4a7e72434e93dfa3ed237fa929 : SUCCESS in 1h 08m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f64fefa75a114f69acda3a1d715348c5 : FAILURE in 20m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ac89bbc3e814eb98880b913ee228fbf : SUCCESS in 1h 13m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0bcb6f13fee9414f9367a27034f6b396 : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98c007e88eb431194892b81c63f903c : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21f616d3977c4f65b5c4fce7913bee85 : FAILURE in 29m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/665bafd445b543eb86b73044f0fb6e32 : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d439ae330c90d0288a26fa43de0cf1deea32cbe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 19:21:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0553da03e2e843d780d08c3280d20081 : SUCCESS in 40m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a52c45f2d1444fefb887637508b3a555 : SUCCESS in 41m 46s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/683373db48d245bcbea9d4df8c830696 : SUCCESS in 1h 25m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/17b2ee333d054e26929e3032822a9a40 : SUCCESS in 30m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/633205a199974d55b365a3e390b04ee3 : SUCCESS in 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2710896e0f04d255767bad85e4d24d0f92acfe22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-18 19:05:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6ff06e19ff89ba113429f41c1853c479e31b1e52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 20:20:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be651d131c2a426e8b77f7abc0f84c14 : SUCCESS in 10m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/4eb2f94d7e8345db818d0d463e815882 : SUCCESS in 47m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ebf1c5b3f74246828bc8bce5be843b2a : SUCCESS in 37m 05s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/586f872787f54a0b8c4dfc9a5a47dc15 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0fbae1476edf42b89adf8a6a1c0bee8a : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03f2a1ee88814943a67a93b7e27e1632 : SUCCESS in 15m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/85e340d4a9ef482e9f8b8443a180aa76 : SUCCESS in 14m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d1ff03fbc7da47caa7ee9e2533c90de7 : SUCCESS in 18m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2b3d432512e4d9bb5e4c4417b95608d : SUCCESS in 16m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/035d454e43dd42bd942c1c4d0963363d : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4bc9a6a686874b058ff035bb8ee63694 : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7c69d411f37a449b95c097cfbca8d23d : SUCCESS in 16m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33f9e24bbee6430bb3dc08d89a8ebc56 : SUCCESS in 16m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/df9737110bba4d879587ed4d224af9ae : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/185532c9b02940f296556642514f065c : SUCCESS in 18m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/df3986874ecc4001be6e0a20d17e357a : SUCCESS in 16m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/921afc34c1ce49509f4104f0061a1c17 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0e1c143e943c483b912a4759f6b07c08 : SUCCESS in 18m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4aaea2a0e6584ab490826ef0485345d2 : SUCCESS in 1h 00m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ff87b09b45924e719dd02161c8c60a5b : SUCCESS in 1h 04m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ba1614333044444b8509e7859e39cfc9 : SUCCESS in 38m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f96ea1b74cac4bb9a612bf3f54c2491e : SUCCESS in 52m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c089b90189bd4bd8869a8fb803ed79c4 : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f9e4cb926454d60a259cd475cc4fe31 : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2265afc1da6645468c776703a8a3adae : SUCCESS in 24m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8f6636c8aa1498caeebc9ddecb8f407 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"9358da3c9663c088cd1be8d1d7c86e9961dc6bfc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 20:52:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d3e3c92f0364102bfec6b6b18d7c797 : SUCCESS in 31m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c04a18d8822d4910ab7c197dbb5e3755 : TIMED_OUT in 1h 31m 37s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/0c714e34d3fb4e969340cac6408d3254 : SUCCESS in 1h 36m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/34e11e820bb24791981be75bdadc2eb9 : SUCCESS in 31m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d96cf94fc0949deb62dd55f4c848dee : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5074b32e725f04e912174f8bf1ae8549078aa520","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-11-18 21:17:22.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"afc9cf5570fa172105bc16ea4b1c1d7e5ae99703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-19 09:27:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ec9c86bc2491de972fe0732d1720cf641620e403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 10:41:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34a7c64457024af3855f9352b5516237 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/e06e70cbaec54e1892da12097e269c49 : SUCCESS in 53m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb5a592068064dedbe550e13eb7b8a19 : SUCCESS in 46m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e307243b9724ef1abdbc714cd1df8ab : SUCCESS in 8m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/54fc7fe2ec3049b38e9d93e585105390 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/21499f2cebcd4e5b99a2e175e7e5dc19 : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7203510520024ef0a230f9c5dd3a2044 : SUCCESS in 16m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0525cd0552f4d17a98e77e885d67844 : SUCCESS in 16m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/99d643902cc44603bf35306a6a17f67e : SUCCESS in 15m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7b6bd08e3037417da4524993152c48d5 : SUCCESS in 24m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6da1446b46fe41bcb072afa2c82c909f : SUCCESS in 25m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28ea3b953b05495e93e31fb71a2c2b21 : SUCCESS in 27m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/841802e4d09b4942b9f21d5767628f1b : SUCCESS in 24m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2408a5cde8e04a5783b12de152dd88d3 : SUCCESS in 22m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2ee40f21a6340ada5cebaaa68d8277e : SUCCESS in 15m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2fb1afb6c0674d98a4f54b68fbb8724a : SUCCESS in 19m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/137bd0d85c1f47c9b0cb615de1b362e9 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b888921a5949448fb5b048a0248d0a8b : SUCCESS in 23m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5023dda0963645c6a9c026d2abcc750f : SUCCESS in 51m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/15f1deb70faa49a1ba7fceb916573a7d : SUCCESS in 53m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2aeb79e41dd46b4a84b07ffb9b03fce : SUCCESS in 47m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7192073815754a808616e106a5fdf700 : SUCCESS in 51m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fcfefb6105e4723b4cd216d81bb4b6a : SUCCESS in 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c17fda32ef874e698e47c7f5de3efb86 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a5b404218214cfab9f78b09e8e69ddb : SUCCESS in 22m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65a8d8f262b6468ba83c486e6b7d9747 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":3},{"id":"e001ab910e280f5695b9683f732fedb0b8b138fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 11:45:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/356f6f23162c44d6b7c163e04a0850c6 : SUCCESS in 54m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c88ccebb6c4458cbd4dbb46ed4f40ef : SUCCESS in 1h 02m 55s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/60d98109e554462d862143493b8ac97d : FAILURE in 2h 12m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d05a3ab6972148a5818df1a6328832cd : SUCCESS in 1h 10m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfe1f3fbfd0f47d89dd7ee936e73cbfa : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":3},{"id":"72b4c6bfc18f367c5926b752d6d3c441e6cd1f3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-23 14:58:30.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"70bd0bf19b5d9b4ce7406fc03a72d1fe08236966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 15:58:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ca3ffc7b5f41a4a4c9e87d74e5cfae : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/f109a524a89143bc9ee821a2e0e24b87 : SUCCESS in 51m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da426dc300994637b1973e9044c41826 : SUCCESS in 47m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8ac3b7df74d24e58b411e7b5c04cc6c4 : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c965c67654c4dc78b50944cece5d92e : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbcc985951c14511b66234873c52e03f : SUCCESS in 13m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3038470c6dc43fa95f38dea9f5fba04 : SUCCESS in 22m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad613f8339d5465093e0a9cc7c04fb9b : SUCCESS in 18m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/53180220049745378338af03e982fdde : SUCCESS in 16m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cd9220498292437593736ed7058431ac : SUCCESS in 14m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/580fedf2d58c467d9ea67630286e084a : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ac852648fcf644839d1c64ff624c2965 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d32d193712f439990c1611aff8bae50 : SUCCESS in 23m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/acf36933bc8744eaaf0f86c9be5be182 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/774b4db33a664a6093468cb516a51ca5 : SUCCESS in 19m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a47e8ae454bc448b8b87086178741b60 : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8539ac224f124f63a355a4c2d0ce3cb5 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8fca660a37644cd399becdb4745ecbe2 : SUCCESS in 28m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/acf45737e672468ca62fb1c21f8a130a : SUCCESS in 43m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/781705583a22450b9029da4579dc4ed0 : SUCCESS in 52m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75c3a395248645a49089611e4821ba48 : SUCCESS in 34m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/63c0caddd8f44de99dde6e834e55727d : SUCCESS in 35m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7ab5c7404644416aabaaa867ee41011 : SUCCESS in 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f86477492149bea30aa97d6419d8be : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1587fc4e1ec4efebb600b58ed792a99 : SUCCESS in 22m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c2b72bb9aa1409cab76f0dc8309f064 : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":4},{"id":"dba1ed147b8dae0df5067a8b69306921e840d1b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 17:06:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff387dc330b4490a9f8f106f18b3fb03 : SUCCESS in 40m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20bba2299720422fa17e32772ef29e4e : SUCCESS in 46m 37s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/cbf57a612def4fe49f1425aaca469a8c : SUCCESS in 1h 38m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cd0c3c6f5ec45d58651cea27100d68b : SUCCESS in 50m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/455cb9edcfac48d8bbb8fec9cc5b4e10 : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":4},{"id":"6bd6907f5ca1bd05146d0668d909307fe355e1d8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-24 12:49:07.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"276e5f1490aa65d40a8f108b7110bb899a31d07a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-24 12:51:32.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9e262ccd36c8b270c964b1708be0660596b25c24","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-24 12:52:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"15cf4ddb270421d8e9c39fecab1effd5880d5f9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 14:09:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/034cdd2b0514433a83a377c38e128a3e : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/77d60d35d7ce4701833fcd844650ef16 : SUCCESS in 52m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5395f1ae6c74473598038d99fd5ebde5 : SUCCESS in 41m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e30bd354ad24414a970050ab3670eab4 : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ecb2199e5b664907a6116af06f945616 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36ff0f3be1fa4e23a5d5be78e83ec860 : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/859c23b6744947d2bcdb37647f3e5bf4 : SUCCESS in 20m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d5f1448580544617940437d4aa7753e6 : SUCCESS in 16m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/69175dfbeaa240c888a511f4b3f715f7 : SUCCESS in 20m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f241c28927754174ab8e7482e106f7e2 : SUCCESS in 19m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/90352bcb90134dfd95c3c120aa340bac : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7490514bb25e4aa8a6195fb2215875e4 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8d42109d2684c36bc89317445dc5938 : SUCCESS in 21m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f49834fea73e47f894324323c2b40484 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8b01eb9a315b47e8b8903520b46d2cf8 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/68aeb3735ab749af8a12e9e5b3e4c4c2 : SUCCESS in 20m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/99cd578f42ad4a748c1aa16f6f85af40 : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af2cf54badc8408eac6240a6feddacb2 : SUCCESS in 24m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/241f6411524b4b7bab6d0309d8e3b163 : SUCCESS in 52m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/82f89c3661d94f53bc1f2be7e1520292 : SUCCESS in 1h 00m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15fa527a2fe54736b6cff763554abd4e : SUCCESS in 43m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5092f4d1a1514bc283d6cba81d5e39d9 : SUCCESS in 42m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90058b139e9f483ea9e8a1455a4c4568 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62abbaf542ee4fb2bcbecf3a8309c054 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8420e83e8e346a89c569114643fa47d : SUCCESS in 26m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68db9b1469424ce0a26a621c9030b606 : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":5},{"id":"875cb2b3f5b3c614f223d2c34b5c8e0041006246","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 15:57:37.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9565962f4a94935a81f2a21397cfcea : SUCCESS in 39m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d4db56aae5845668dd91703df4d7dfe : SUCCESS in 55m 40s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/b4d95ea28c4c4159be6579e8c89dc71c : FAILURE in 2h 18m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec53332a3dcf443a89a26b4c0da3b25b : SUCCESS in 35m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ec3a06541fb40ceb6151abd9582a8b9 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":5},{"id":"676c13f7f49c5d989ea409cfc451c27f289ca262","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-24 18:20:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d9515e9743a8109334ace410116b49a5449cc447","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-25 23:08:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"79e339d991ca39c0b928ffa89c7d82ed47476629","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 23:08:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"401f3d3f89f65d394fc7c1ed2915127f60f333e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 00:05:18.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0c712066f749c9b339958515680604 : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/508d7b60978d42ae8bfa131c67187118 : SUCCESS in 40m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/784236166f414443b3b02c228f53288a : SUCCESS in 32m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6774f69dd2af47e7a91086da682fe1ef : SUCCESS in 7m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/dc44780193554964b291e663e822a5cb : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e90885e4623549bb9ccf59577e7fe3ca : SUCCESS in 14m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c224259a46a4ab78e30a52ea4ecdba3 : SUCCESS in 14m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/93ac492f3a0e4a1a84ec6d74794e8345 : SUCCESS in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0a790f06f48d4b64b42cc15c28867260 : SUCCESS in 14m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b4478f4189ad4c8a865502867bfb00ed : SUCCESS in 14m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0e376ea3b034114a55b0b3d0496cf90 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6548a6cab0a14d1cb6abcc6f426f838a : SUCCESS in 16m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9fdb82d7850a4c8e9bf85f265c49cdde : SUCCESS in 21m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b223be90eccc42e099f2b66da319df03 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dda1300edd084ae499a2f48c01c546c5 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d7971a3b213047b7aaeea94f94b4480a : SUCCESS in 18m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a30588bb5eb7478381d9688c7939f524 : SUCCESS in 5m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b7c6a12925634766ae704f08e850aa33 : SUCCESS in 53m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/381d90bc03524a77b1dd581ff9f16ce5 : SUCCESS in 55m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbfc0b9474d549909d9cbc1ebf6ecfff : SUCCESS in 30m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc17eb82ded24e0eb773ca1d5d5d0356 : SUCCESS in 26m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed413de276504aa48cbbcea910af02ed : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf036afed03246f5a7290c07b39f7ef9 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a0b8c28f3a43423e9d5153f5f56d3681 : SUCCESS in 22m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ad1c42a041c4285a9d9bd4cd520068b : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":5},{"id":"ee2ad10f892b508f5bd15c739ecf9f7e0033517b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 00:05:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b7a56753e4ece3211e0e16ecfa8392743b4bc10e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 00:09:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0df0b5395d5b4894a7fe466799f2f460 : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/eff338559ff14fe9b131bc367ac2f614 : FAILURE in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2736624a7ed2463e94ca9c4c209e5437 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"346f518d6cd10b62c411fa2b5fb480b81905568a","revisions":{"2deef46319a0b890c40eebc6e272e7fe078e3121":{"kind":"REWORK","_number":1,"created":"2021-11-16 17:46:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/818127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/818127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/818127/1"}}},"commit":{"parents":[{"commit":"32abcacbff30208cff12211e57af1fed978a1c3f","subject":"ssync: restrict max objects per sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/32abcacbff30208cff12211e57af1fed978a1c3f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-16 17:38:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-16 17:38:09.000000000","tz":0},"subject":"Make arm jobs voting (but not the pipeline)","message":"Make arm jobs voting (but not the pipeline)\n\nMake the non-voting pipeline indicate failure when any of the jobs\nfail.\n\nChange-Id: I84f59256df3aa41338df9829a565ae830ee8e847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2deef46319a0b890c40eebc6e272e7fe078e3121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2deef46319a0b890c40eebc6e272e7fe078e3121"}]},"branch":"refs/heads/master"},"0a6c56e34f278b71bd1ce9cf175258afe6623433":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-18 19:05:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/818127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/818127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/818127/2"}}},"commit":{"parents":[{"commit":"5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a","subject":"ssync: restrict max objects per sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-16 17:38:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-18 19:05:18.000000000","tz":0},"subject":"Make arm jobs voting (but not the pipeline)","message":"Make arm jobs voting (but not the pipeline)\n\nMake the non-voting pipeline indicate failure when any of the jobs\nfail.\n\nChange-Id: I84f59256df3aa41338df9829a565ae830ee8e847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a6c56e34f278b71bd1ce9cf175258afe6623433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a6c56e34f278b71bd1ce9cf175258afe6623433"}]},"branch":"refs/heads/master"},"f9fe0191095c19f11991715d955a5eab13b8c04c":{"kind":"REWORK","_number":3,"created":"2021-11-19 09:27:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/818127/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/818127/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/818127/3"}}},"commit":{"parents":[{"commit":"5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a","subject":"ssync: restrict max objects per sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-16 17:38:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 09:27:14.000000000","tz":0},"subject":"Make arm jobs voting (but not the pipeline)","message":"Make arm jobs voting (but not the pipeline)\n\nMake the non-voting pipeline indicate failure when any of the jobs\nfail.\n\nChange-Id: I84f59256df3aa41338df9829a565ae830ee8e847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9fe0191095c19f11991715d955a5eab13b8c04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9fe0191095c19f11991715d955a5eab13b8c04c"}]},"branch":"refs/heads/master"},"5e4b6add0c8be9959b6d073b5d05800f978f6bdc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-11-23 14:58:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/818127/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/818127/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/818127/4"}}},"commit":{"parents":[{"commit":"a758d7acba11ef204eaecc2ecdf8e4140acb7e18","subject":"Merge \"Fix SSYNC/missing_check blocking (sender)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a758d7acba11ef204eaecc2ecdf8e4140acb7e18"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-16 17:38:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-23 14:58:02.000000000","tz":0},"subject":"Make arm jobs voting (but not the pipeline)","message":"Make arm jobs voting (but not the pipeline)\n\nMake the non-voting pipeline indicate failure when any of the jobs\nfail.\n\nChange-Id: I84f59256df3aa41338df9829a565ae830ee8e847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e4b6add0c8be9959b6d073b5d05800f978f6bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e4b6add0c8be9959b6d073b5d05800f978f6bdc"}]},"branch":"refs/heads/master"},"346f518d6cd10b62c411fa2b5fb480b81905568a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-11-24 12:51:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/27/818127/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/818127/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/818127/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/818127/5"}}},"commit":{"parents":[{"commit":"a758d7acba11ef204eaecc2ecdf8e4140acb7e18","subject":"Merge \"Fix SSYNC/missing_check blocking (sender)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a758d7acba11ef204eaecc2ecdf8e4140acb7e18"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-16 17:38:09.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-24 12:51:32.000000000","tz":0},"subject":"Make arm jobs voting (but not the pipeline)","message":"Make arm jobs voting (but not the pipeline)\n\nMake the non-voting pipeline indicate failure when any of the jobs\nfail.\n\nPreviously zuul would report \"Build succeeded (ARM64 pipeline).\"\neven when individual job(s) had failed. Now zuul will report\n\"Build failed (ARM64 pipeline).\" but the pipeline still does\nvote on the patchset.\n\nChange-Id: I84f59256df3aa41338df9829a565ae830ee8e847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/346f518d6cd10b62c411fa2b5fb480b81905568a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/346f518d6cd10b62c411fa2b5fb480b81905568a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
