)]}'
{"id":"openstack%2Fswift~818388","triplet_id":"openstack%2Fswift~master~I9359feedd93cec482dcd575800e28850c0fc02f3","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2026-06-29 15:49:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9359feedd93cec482dcd575800e28850c0fc02f3","subject":"Add [filter:tempurl]/path_prefix configuration option","status":"ABANDONED","created":"2021-11-18 14:03:32.000000000","updated":"2026-06-29 15:49:31.000000000","total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"95c47587bf048c2af23ee0833984f47b2c169069","_number":818388,"virtual_id_number":818388,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-18 15:58:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"disliked":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":-1,"date":"2021-12-20 22:04:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2021-11-18 15:28:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 15:28:40.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-11-18 15:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 20:47:36.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-12-20 22:04:19.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"c347d51a6de1eb0b035da2d14677d65140566ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-11-18 14:03:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f69e967eb822d8146dd6dccef34719d299377203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-11-18 14:21:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41dce4fb9c8c49416aa77e25107dd9496850947f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-18 15:28:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"252763bed4429110f38b49d76c6bf377c2393396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 15:58:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5f01fd194e4bd0b030c0548ebdc3a6 : SUCCESS in 8m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/18ecb58967344720aef5d59ab466b8f1 : SUCCESS in 59m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca90a434c4f949f58a3815473274a998 : SUCCESS in 41m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/837037f7be974dc9a6953d8c3cce26a1 : SUCCESS in 7m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5377efbed1b34f66b5fea145a9ce3ca8 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7b33394bf47482ab5d8013674030906 : SUCCESS in 15m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c38d6bf3faa1492fa9d8c62054bca44e : SUCCESS in 12m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dc058bcd3da648b2871d10a6ceaf5e03 : SUCCESS in 13m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c31d5d1917a848eb9b219b099a353166 : SUCCESS in 13m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/616177678ec44b33a1985591814be9a2 : TIMED_OUT in 1h 31m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/72b239ef2b0b45d9866db6345a91e359 : SUCCESS in 19m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d54db9c7c6114e6182f731f2bbfd3804 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ba67cda40d74c2b9d7ecaaf58fe7dfa : SUCCESS in 20m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/031fd64c23564cd48429580ddfb008fd : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3709185befdd478895445bd877eda192 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e18ce4914fa048c092e7e5a207306c53 : SUCCESS in 39m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/488ff43d48494018aa48bcff798113d4 : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/debe7ecc93c14684a34f5a571e7c4e8f : SUCCESS in 20m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/40cebc1e9f0c4426b4cdf87eea39474c : FAILURE in 56m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/428d9d3a00a24f4a8bf52d34a3491a06 : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/444c56e04b8c4f81a2a01bbf0e5054b0 : SUCCESS in 46m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4eb183bba304056a7d0860ff2b32052 : SUCCESS in 30m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af187a48dbd5444c87336ac16ea27e45 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf22ec6dbfb7414c95acc1907090a033 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68c0d6dcfdf242de8c343da8223ca5a3 : SUCCESS in 20m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fbeed25065d430a8511232d295037cf : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3d9f22aa1929ebe2cc47e77807008c9199d15efe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:10:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f12ca2a317fe4db2bf8481b490419da6 : SUCCESS in 35m 05s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/282ed3b0791144db844304b7b48c7399 : SUCCESS in 31m 16s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/a5cdf7a2d0184dfd8529b24580df42c7 : SUCCESS in 2h 46m 30s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/24460a271e2440898a9b3cd75d7ff418 : SUCCESS in 56m 59s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/14ae11caa85a4266b23ef027d5ddf679 : SUCCESS in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2d678a02951f20470263fbe8fc0d1b907397d9a9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-18 20:47:36.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dbffe5771d58dff0debfe8bcb4beda92557c27f0","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-12-20 22:04:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95c47587bf048c2af23ee0833984f47b2c169069","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2026-06-29 15:49:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"836e425387f8869b032dd51eb8c93a5bcb0d5f90","revisions":{"0b15daaa5367b019113ed6a94b86aebef1b5ae1d":{"kind":"REWORK","_number":1,"created":"2021-11-18 14:03:32.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/88/818388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/818388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/818388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/818388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/818388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/818388/1"}}},"commit":{"parents":[{"commit":"f91074f376937699dcb0901a9c46030032730773","subject":"Merge \"Make cmp_policy_info agree with the API behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f91074f376937699dcb0901a9c46030032730773"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 14:01:45.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 14:01:45.000000000","tz":60},"subject":"Add [filter:tempurl]/path_prefix configuration option","message":"Add [filter:tempurl]/path_prefix configuration option\n\nIf swiftproxy endpoint is something like /object, with URL rewriting\nby haproxy, then the hmac calculation is wrong.\n\nThis patch adds a new path_prefix directive which is stripped away\nin the URLs before calculating the tempurl hmac.\n\nChange-Id: I9359feedd93cec482dcd575800e28850c0fc02f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b15daaa5367b019113ed6a94b86aebef1b5ae1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b15daaa5367b019113ed6a94b86aebef1b5ae1d"}]},"branch":"refs/heads/master"},"836e425387f8869b032dd51eb8c93a5bcb0d5f90":{"kind":"REWORK","_number":2,"created":"2021-11-18 14:21:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/88/818388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/818388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/818388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/818388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/818388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/818388/2"}}},"commit":{"parents":[{"commit":"f91074f376937699dcb0901a9c46030032730773","subject":"Merge \"Make cmp_policy_info agree with the API behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f91074f376937699dcb0901a9c46030032730773"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 14:01:45.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 14:21:17.000000000","tz":60},"subject":"Add [filter:tempurl]/path_prefix configuration option","message":"Add [filter:tempurl]/path_prefix configuration option\n\nIf swiftproxy endpoint is something like /object, with URL rewriting\nby haproxy, then the hmac calculation is wrong.\n\nThis patch adds a new path_prefix directive which is stripped away\nin the URLs before calculating the tempurl hmac.\n\nChange-Id: I9359feedd93cec482dcd575800e28850c0fc02f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/836e425387f8869b032dd51eb8c93a5bcb0d5f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/836e425387f8869b032dd51eb8c93a5bcb0d5f90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
