)]}'
{"id":"openstack%2Fswift~818400","triplet_id":"openstack%2Fswift~master~Id0149252ac0336c6551d06369dfd72074d85b79b","project":"openstack/swift","branch":"master","topic":"collections.abc","attention_set":{},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2022-01-14 20:03:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id0149252ac0336c6551d06369dfd72074d85b79b","subject":"Fix import Iterable and MutableMapping","status":"ABANDONED","created":"2021-11-18 15:17:26.000000000","updated":"2022-01-14 20:03:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"30965b5d853bb995427a9fcd3d9a53746547a7bf","_number":818400,"virtual_id_number":818400,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-18 17:00:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 17:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25a93ce04ee5afe4bffed04a2d7c349c815bd1c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-11-18 15:17:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f4c648bc8e11d68bb0869f836cfdc0d5e93879d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-11-18 15:17:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee6b7b9431cd738b9468c30ee87e1c67233268bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:00:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488d67642ec64a45853c2dcb03030d09 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/a95b1752acee46d1a59bc71d364084c2 : SUCCESS in 1h 01m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa32d6b4b24b4af3b51dae753de78252 : SUCCESS in 46m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c79e5789c8fe4f00966bbed086681168 : SUCCESS in 10m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/86d0185dd4ee48fcb650261360cc487d : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/926c9045ae424b60b595be5d42b75c9a : FAILURE in 5m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4313cee9cc5a4922ac0cf0d138c9218a : SUCCESS in 24m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3da4f773fcce49fba5204971f59b1f2c : SUCCESS in 15m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d996e73fd384bc999e47f2e94fb685d : SUCCESS in 17m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb4571923d58436cb4426e307134514a : SUCCESS in 17m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/500ee4c749824ba8b32d9a05dd4bd88b : FAILURE in 6m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9a94e0058e1c498f8ef2135487f52271 : FAILURE in 4m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4feec112ac0d46e89a4d3ecd7acf72cd : FAILURE in 6m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f375fdf0ef9b412380f231f1945e77d9 : SUCCESS in 21m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f30334a169e040edb7c73e19c7c05539 : SUCCESS in 20m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/761fe04585bc48c696ecfb8a5b157c5b : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/659cc4132da84ba6979244627bcca4a4 : FAILURE in 7m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f7f241ec7c43498690be79bdb958aa96 : FAILURE in 5m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/414523c159dc4359b810a3826bd4b560 : FAILURE in 6m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/75237a02c54a418986f6a32e1fc8b128 : SUCCESS in 54m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4b86b914a3fb4262aa7fb036dfe6156b : SUCCESS in 45m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e00d00d5cf0430089650d40f36e9529 : SUCCESS in 44m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07ca373b200e4ef9bd4b47fb7c04c020 : FAILURE in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f5b173351448c7b641fd74eb497656 : SUCCESS in 7m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a00462ab5c20468badec41f27c8144a4 : TIMED_OUT in 1h 31m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/136d3846eace41678af85b3cecd83074 : SUCCESS in 58m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e1d35b939d7fcac263d9586ed6d0594fe3c24929","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:11:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c8fde6cf1e4428e8fe4c9993e73af57 : SUCCESS in 40m 44s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb295a8b3c16423aac8d97b76a2e00f5 : SUCCESS in 40m 28s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/4727706a670f4f29bf9b8f9e96707c2e : SUCCESS in 1h 44m 55s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc92d1c39c1e4f9ca17f4efc4837841e : SUCCESS in 46m 09s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/65bc5d66bd4042d0bddf38f84f9e6cda : SUCCESS in 57m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30965b5d853bb995427a9fcd3d9a53746547a7bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-14 20:03:02.000000000","message":"Abandoned\n\nFixed in https://review.opendev.org/c/openstack/swift/+/814823","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"991ba99e6df6f454d758f27fd433b73a5fa73551","revisions":{"0780d0c3cccb6dc177e36deeb9d7081f94388b00":{"kind":"REWORK","_number":1,"created":"2021-11-18 15:17:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/00/818400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/818400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/818400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/818400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/818400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/818400/1"}}},"commit":{"parents":[{"commit":"f91074f376937699dcb0901a9c46030032730773","subject":"Merge \"Make cmp_policy_info agree with the API behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f91074f376937699dcb0901a9c46030032730773"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 15:00:14.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 15:17:12.000000000","tz":60},"subject":"Fix import Iterable","message":"Fix import Iterable\n\nSince MANY years, the import should be:\nfrom collections.abc import Iterable\n\nand not:\nfrom collections import Iterable\n\nSame problem with collections.abc.MutableMapping\n\nThis patch fixes it, so swift unit testing can work with Python 3.10.\n\nChange-Id: Id0149252ac0336c6551d06369dfd72074d85b79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0780d0c3cccb6dc177e36deeb9d7081f94388b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0780d0c3cccb6dc177e36deeb9d7081f94388b00"}]},"branch":"refs/heads/master"},"991ba99e6df6f454d758f27fd433b73a5fa73551":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-18 15:17:57.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/00/818400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/818400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/818400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/818400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/818400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/818400/2"}}},"commit":{"parents":[{"commit":"f91074f376937699dcb0901a9c46030032730773","subject":"Merge \"Make cmp_policy_info agree with the API behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f91074f376937699dcb0901a9c46030032730773"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 15:00:14.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-11-18 15:17:44.000000000","tz":60},"subject":"Fix import Iterable and MutableMapping","message":"Fix import Iterable and MutableMapping\n\nSince MANY years, the import should be:\nfrom collections.abc import Iterable\n\nand not:\nfrom collections import Iterable\n\nSame problem with collections.abc.MutableMapping\n\nThis patch fixes it, so swift unit testing can work with Python 3.10.\n\nChange-Id: Id0149252ac0336c6551d06369dfd72074d85b79b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/991ba99e6df6f454d758f27fd433b73a5fa73551"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/991ba99e6df6f454d758f27fd433b73a5fa73551"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
