)]}'
{"id":"openstack%2Fswift~818466","triplet_id":"openstack%2Fswift~master~Ia11f892ccfeb54158c5e68746c799f1008057ffc","project":"openstack/swift","branch":"master","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-11-18 23:10:54.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia11f892ccfeb54158c5e68746c799f1008057ffc","subject":"DNM: try to force ssync into a missing check stall","status":"NEW","created":"2021-11-18 21:03:10.000000000","updated":"2021-11-18 23:10:54.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"22dfa0fd2178383b8bf2dce1652b61eec6791638","_number":818466,"virtual_id_number":818466,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-18 23:10:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 23:01:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-18 23:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f12b6a730b2847c4c788a77322eb3bca94d98b1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-18 21:03:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d63d357ee0df239fa4fb4d9066333b597c021a4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 23:01:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f426ceefa7aa4e828ceadcdcd918e5e9 : TIMED_OUT in 1h 31m 19s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d30380a85d1647fbbd9ddb42ae87b47c : TIMED_OUT in 1h 31m 28s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/7970edf92c6b450db9475c8d27cd6bdf : FAILURE in 1h 53m 00s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffc840ec929b488f80f29762b27d98dd : SUCCESS in 33m 13s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd116ad27f0847c7b14b0bbb2600211f : SUCCESS in 58m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22dfa0fd2178383b8bf2dce1652b61eec6791638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 23:10:54.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ab5e2c7d784093a6a50ef466d8c4a6 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/8e3fcb95e7ad4e51b8c097cca11208e7 : SUCCESS in 1h 11m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3086d79176f94fd39dcde8188269337f : SUCCESS in 44m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9434137a32048fbb9f3ca1ad5ccc62f : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/43e78a1f52a9467c875fda447ee6f144 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a131c70ff3b04b2fae107b32d0045021 : TIMED_OUT in 1h 31m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f67e1ea135234662bffd49bcf1fddb1a : TIMED_OUT in 1h 31m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50653aa844584e4caa46b9b7f60c0f5f : TIMED_OUT in 1h 31m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be830ff5dc774e04acb1fe874cdc18b2 : TIMED_OUT in 1h 31m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab1f2a7148f2425196442efa89d09783 : TIMED_OUT in 1h 31m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b13f8b5bbc204ece8bf15283ed378b00 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1ba166de4f7444e8a069f96b9ab0639 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee5b1d8f935a4349ac7721cff54b1236 : SUCCESS in 24m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ce487ab7aa984bd2b36de3c118a00e45 : SUCCESS in 25m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f1f719190dc84827bf668195edafc7f4 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f88e70294a1d4e1dacc591ac0b48f41d : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6585ab5d00184312b19b9595114d9d70 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81facb57af524ed9aae337d3a70c8b57 : SUCCESS in 22m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d89809ac8d444058936e13371a70f5d : TIMED_OUT in 2h 01m 16s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6800ae25a7c14b3482cc5a08ee2c8cc0 : FAILURE in 1h 30m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ec20c1cb8934209a8211628ec134f25 : SUCCESS in 42m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20c19f28c2f04e3eba3e74cae50b8105 : SUCCESS in 39m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a12f276a5774e7286e345c15a2107c1 : TIMED_OUT in 41m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa458f833204eb7b997e01e207ca2e9 : FAILURE in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f89f5f927cc4d37a71d128afd09a54c : FAILURE in 11m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2227329279445baaedf173930affecf : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"454ba815bd2c1a4c4995676c2f27ae3bca6eb6b5","revisions":{"454ba815bd2c1a4c4995676c2f27ae3bca6eb6b5":{"kind":"REWORK","_number":1,"created":"2021-11-18 21:03:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/66/818466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/818466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/818466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/818466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/818466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/818466/1"}}},"commit":{"parents":[{"commit":"5bca838018dd7cfc2b60a38b89d91196a7a46a86","subject":"DNM: playing with ssync EAGAIN","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bca838018dd7cfc2b60a38b89d91196a7a46a86"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-18 18:32:17.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-18 21:00:10.000000000","tz":0},"subject":"DNM: try to force ssync into a missing check stall","message":"DNM: try to force ssync into a missing check stall\n\nOnce ssync sender connection.send blocks, it does not appear to be\nunblocked ny the receiver starting to read.\n\nOnce the sender timeout fires, the receiver will continue to read\nuntil it reads a malformed line because chances are the sender sent\npart of a line before stalling, and it\u0027s not going to send the rest\nonce it has timed out!\n\nChange-Id: Ia11f892ccfeb54158c5e68746c799f1008057ffc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/454ba815bd2c1a4c4995676c2f27ae3bca6eb6b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/454ba815bd2c1a4c4995676c2f27ae3bca6eb6b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
