)]}'
{"id":"openstack%2Fswift~818550","triplet_id":"openstack%2Fswift~master~I5dfdf5950c6bd7fb130ab557347fbe959270c6e9","project":"openstack/swift","branch":"master","topic":"p-reconstructor-purge-meta","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-11-24 12:39:38.000000000","reason":"Alistair Coles replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-11-24 16:55:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5dfdf5950c6bd7fb130ab557347fbe959270c6e9","subject":"reconstructor: silence traceback when purging","status":"MERGED","created":"2021-11-19 13:06:23.000000000","updated":"2021-11-24 17:01:45.000000000","submitted":"2021-11-24 16:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"818550-p-reconstructor-purge-meta","meta_rev_id":"10d595fe3d1e091a0a80bc4d4b3ed66f80fdf597","_number":818550,"virtual_id_number":818550,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-24 16:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-11-24 12:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2021-11-24 17:01:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-11-24 12:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 14:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 19:09:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-11-22 16:17:11.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"c9cd34781f63784b1f97df67f19ab99c499fe138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-19 13:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a1a56ee0f2c48d3368230c0c47b2fcf9deccb0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 14:22:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc40485afb944d0a546a109bb444d85 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/f7703e27b23541ab8700bd5c21356abb : SUCCESS in 58m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fbfad1bbf591490492584ea38dca4120 : SUCCESS in 51m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7aa0ba020c964c0bb28be8bce55684b3 : SUCCESS in 10m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb705e46d5fd4a6ebedd6964ff9a313b : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37057fd3bf7f4989ae178163952f2cb4 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61c4ac9ec0bf4b758bcab18163b0af55 : SUCCESS in 15m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d356640e8502439da31f72a106bbfc83 : SUCCESS in 15m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68485a67b7b74f1b8dd2d645aee205ce : SUCCESS in 16m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f1972accaf34715bd559f92391282f3 : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/136462576c434b8b85fdad1ebc33161d : SUCCESS in 20m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/15350ebd21704b35ad8e38bf807b4b4b : SUCCESS in 18m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0a2443c061854582a1f06bc6291c401b : SUCCESS in 23m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9448a61127f24565aaad10144ca1cc0d : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9969b0dedc854fa683f47a4ddd926ab3 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/23ad7e999d0d4f648adba15078437053 : SUCCESS in 20m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c66fb580953c4c1ba658b0f8541716ee : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9bf9efb0008497a9ca8c6e4eba63776 : SUCCESS in 20m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1f79612bdda34a8ea9ba59e2cfc24ca2 : SUCCESS in 51m 36s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4e3679bc8d3740179fb77e1dfe6ba7c5 : SUCCESS in 1h 08m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90445f957f80444688d5d089bee423be : SUCCESS in 39m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c1c0fb9d5e747bfb78e569e1bf6df0e : SUCCESS in 38m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7355cc27e5545af9ff9c6631af5a55a : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288b54cae7f34c4a81e0befc2ca5809d : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de509dd2fe36418ba48b69a092d6b620 : SUCCESS in 22m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f61cd19b8ae4e5caa083c9e4850aa4c : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a9d279d44c25474455abf9370dfea42ee5f82ef5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 14:52:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/179ac6927f344a0a8b6544c56af7ae4f : SUCCESS in 42m 36s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2061789b6a2f499db054a3e25a9af2dc : SUCCESS in 40m 05s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/ab0a3b88c6024d7c958a504696183995 : SUCCESS in 1h 37m 51s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8005621ce604c629a2546c557b73368 : SUCCESS in 36m 19s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6d88ced28d9489e88250ea83b2cbdbb : SUCCESS in 36m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32fe57bc44d9e50e138d2b085e4ecc665b211d6c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-19 19:09:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9dc7aff2a1bf86a42c37f56d0e0aa81a96ff038b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-22 11:55:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca25c09cf6637e1f135c86418ad20ddef29b35fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 12:56:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b4fa4b4d4c401596fb69b71ed39c42 : SUCCESS in 11m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/dcb160bbeded4fceb82407d97a83111e : SUCCESS in 53m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e29629fd4dfb4f0b89b048945b47f071 : SUCCESS in 47m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/512c7f7a06ae47acb8ff878dc88991a5 : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7620f71b6003443eac2a4ea9170d55fc : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2f2e3f739cda41278071453258a7ed52 : SUCCESS in 18m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd4e322705ca4d7195b2cf89946bdb7f : SUCCESS in 17m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc3628ccc5ed4baea1df4e384c19595d : SUCCESS in 16m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a18211eadc384b8abb9d02f12470db38 : SUCCESS in 15m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f16bfbdbe8f46b4bdee1afa0bb18fba : SUCCESS in 18m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f5b0104761f48d69ee3fecb746e89ff : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fd35f24933c449d96546771ea8fc8da : SUCCESS in 22m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac3ffe32e21140f1807790e867ef5272 : SUCCESS in 24m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8573a62850f64118a881a6a6fea97c86 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77bdeabe3b494ff28902b606a5676f51 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/83c0d2e8df77404692235d2649e1cc08 : SUCCESS in 19m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b1a95a5be4d464eaca930d8f1a59645 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/595cfb0ada5945e1bdc33379cdb4a345 : SUCCESS in 21m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3f4163eb745c44dd8fa2f3eb2c3f8041 : SUCCESS in 48m 22s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4fbe9000545e4cfebfaa88f7e8cfff25 : SUCCESS in 54m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86f554b8018b4deb8cb7dcb48d68bd94 : SUCCESS in 47m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5677aa5370454aa8bda9f85ac09fa826 : SUCCESS in 40m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c09df0b6c684334a4df99b9decc63b7 : SUCCESS in 18m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca9092632944c0385066016c2e04cb8 : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fca12c6573745d3b9751809256018d4 : SUCCESS in 22m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1186bf9c317b476ba3d0937f3f415652 : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":2},{"id":"91b13620ce50c489e56c78a1abe40a4257f40cda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 14:04:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2131da5e173d4cba8f297ae923c40361 : SUCCESS in 32m 12s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af594fc8cb864e968900951a5e941694 : SUCCESS in 43m 34s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/5289f007b29c4c5b88a773fe33b409f2 : FAILURE in 2h 06m 27s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1742d35f9090474383d615b83302a9c7 : SUCCESS in 1h 01m 00s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc2c310011b64d51804f208caedc4e55 : SUCCESS in 31m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9445a03a6f02026ec24190216b42977e98be9a98","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-22 16:15:01.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"110f028a72a96f0eb044ba298e18fe2cd5a7ec71","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-22 16:17:11.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a63cc37b5eb67f8525762dc2444af8e885249ff","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-23 20:57:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d91e6e28e3270978edab77e32f25c52f7ab345f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-24 12:30:06.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"455d7e480af61a93fe63d1ceb2ad73b4d42c8b02","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-24 12:39:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"29fe511be6b0e67976fc674983942ae5231829d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 13:52:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49dad9264730444896a1b0d35322c4e2 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/dfcfdc6637ab4c73b47a84dbdf96ed45 : SUCCESS in 1h 10m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9460905a06b944ae930e7aa1f1cd9c5e : SUCCESS in 38m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3b167150f304e9baf9638486da4b819 : SUCCESS in 10m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/50a339b879dd45da8745170b760bdfcb : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6be2c5cbc1a4303ab2c563e903cf83e : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1c53f9f4d584ab9b90ba4791467ea78 : SUCCESS in 13m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7dd99e0b248d4b6e9024d827ada4927f : SUCCESS in 17m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ee897e9e4a7e4c358b89ecd494d521a5 : SUCCESS in 29m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/903f0ff9e6854742948a44889545793d : SUCCESS in 13m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6bc87caa8d154f2ebb21bb7173e662c9 : SUCCESS in 42m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a018e7be9a954528aec231d0175a6926 : SUCCESS in 43m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2384c847e8f4868914c24d4da60e369 : SUCCESS in 16m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e3b4dc081ec0420eb6d519079c1594a5 : SUCCESS in 25m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3883947c0711477abe5a8d077e074f34 : SUCCESS in 12m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e56c5487860f4351bd288a10a23fa486 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/507dee9c1a1b49c3b329879874e2c953 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32d694ebb5d94d55b41baa913861bbce : SUCCESS in 52m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93d8eeff41f442b28bc927a343238ecb : SUCCESS in 1h 17m 40s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e9c70c7475b84719bf38619358e713c0 : SUCCESS in 51m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14efdc737e6b4176943badfdd3688a8d : SUCCESS in 28m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/627562abc3cc4ac0b7712794d24ae842 : SUCCESS in 35m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/996333c687fa4577a7418320c367c56a : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6b45ebe5ad479a8a3e6b92d70b5fc2 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/adb7c6a22c5e492ab4ac47d1ad430ac9 : SUCCESS in 19m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61ecac2e49104bcea4f2b5d16eb26086 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":3},{"id":"826ff7ced721450660ce0c48b2eb34ba5e256a76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 13:54:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f0ec1e0765c51d420af7fe0283161377c09ccf47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 14:14:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f939b848b89c471caa442817f50d6a5c : SUCCESS in 45m 48s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84c188a0100a4118a1aaa15cbffd5eec : TIMED_OUT in 1h 31m 08s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/f0e6983ed5cf48d3b734199da91f4b36 : SUCCESS in 1h 37m 33s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b36d0507b2447ebb0ddff573fce155a : SUCCESS in 39m 22s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a7e1bc5c7374b3ead6a372ae7860689 : SUCCESS in 1h 04m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cbf57cf445b52adb1b3e2293a23b17eeef5d8582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 16:55:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eca7f0372110444e817e1e3f1a6ac211 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/c20b1a61b2ca49ad8f760f21a14a3185 : SUCCESS in 1h 08m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/471ffad7cd90442c8ca36df64fa100d5 : SUCCESS in 54m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/daaec1dc819a480ea2563f5eda9e093b : SUCCESS in 9m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e098227907924883aea714d8e78fc1d7 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd7a0a48ec3d4d22a4a5998577d59e22 : SUCCESS in 16m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73de4826bd8d43229ba103aeb280f1d2 : SUCCESS in 17m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba03764745184c1d90887aaf1cb85f88 : SUCCESS in 28m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c604f1b583304b068eecf55c2333e898 : SUCCESS in 16m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b15369a6d22543d4ab42b824db82a5df : SUCCESS in 14m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/929086baf95f49dab58503f6883b90b0 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f74f845078848ccaadc47ff41970203 : SUCCESS in 21m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/767e2405b978483e8edbb7fd7fe1acc0 : SUCCESS in 20m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6a4f4824aa7348d5807a6e497994a7c7 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1456084cf85e4d51a37caef20f6c1bad : SUCCESS in 34m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/743e65f874a644fe9e32013580ec37d0 : SUCCESS in 16m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/15a5ffc17ac043cb86d32d36498525a8 : SUCCESS in 6m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9756dd043c784ba980061e23c85a83d0 : SUCCESS in 1h 00m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bc3b06675ed644b4bcf1ed47d53172a6 : SUCCESS in 58m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48085d3c3b884f1ba37002e936ddbb59 : SUCCESS in 34m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/390d57d012894219850f814695c647ae : SUCCESS in 40m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9d76e21d9184636bf5f94e57512beb3 : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237e07fc51c34ac5ab17757d2557a2c8 : SUCCESS in 7m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6950cad001f2460c96f853d5a14fe916 : SUCCESS in 40m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3118ffb013184720a269e06fa06de3d8 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":3},{"id":"ebe1cd49afc9d2c6e1287dd0586e4b0f6752d477","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 16:55:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9589b7268d7dc5f516a6f13e43173032dea0bff8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 17:01:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7b71b6caa0e491f8602a26be03a22f9 : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f60055ab26594e6e9861993519ae418a : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/2f13d599fddb4646b931e8ce93ff9595 : SUCCESS in 54s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"092d409c4bbe542f6b9f3c217aa05ffd8669c815","revisions":{"8b9222d7a68aad9207f0b1baa07d659d16ae71eb":{"kind":"REWORK","_number":1,"created":"2021-11-19 13:06:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/818550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/818550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/818550/1"}}},"commit":{"parents":[{"commit":"f91074f376937699dcb0901a9c46030032730773","subject":"Merge \"Make cmp_policy_info agree with the API behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f91074f376937699dcb0901a9c46030032730773"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 12:55:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 12:55:59.000000000","tz":0},"subject":"reconstructor: silence traceback when purging","message":"reconstructor: silence traceback when purging\n\nCatch DiskFileNotExist exceptions when attempting to purge files. The\nfile may have passed its reclaim age since being reverted and will be\ncleaned up when the reconstructor opens it for purging, raising a\nDiskFileNotExist. The exception is OK - the diskfile was about to be\npurged.\n\nChange-Id: I5dfdf5950c6bd7fb130ab557347fbe959270c6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b9222d7a68aad9207f0b1baa07d659d16ae71eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b9222d7a68aad9207f0b1baa07d659d16ae71eb"}]},"branch":"refs/heads/master"},"f0e0b4eb1e90b4441b57621afc82c22776d16430":{"kind":"REWORK","_number":2,"created":"2021-11-22 11:55:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/818550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/818550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/818550/2"}}},"commit":{"parents":[{"commit":"5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a","subject":"ssync: restrict max objects per sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 12:55:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-22 11:52:50.000000000","tz":0},"subject":"reconstructor: silence traceback when purging","message":"reconstructor: silence traceback when purging\n\nCatch DiskFileNotExist exceptions when attempting to purge files. The\nfile may have passed its reclaim age since being reverted and will be\ncleaned up when the reconstructor opens it for purging, raising a\nDiskFileNotExist. The exception is OK - the diskfile was about to be\npurged.\n\nChange-Id: I5dfdf5950c6bd7fb130ab557347fbe959270c6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0e0b4eb1e90b4441b57621afc82c22776d16430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f0e0b4eb1e90b4441b57621afc82c22776d16430"}]},"branch":"refs/heads/master"},"092d409c4bbe542f6b9f3c217aa05ffd8669c815":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-11-24 12:30:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/818550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/818550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/818550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/818550/3"}}},"commit":{"parents":[{"commit":"4c84f615a100707981d0faa4d7e325135920bada","subject":"Merge \"Fix SSYNC update phase blocking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c84f615a100707981d0faa4d7e325135920bada"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 12:55:59.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-24 12:11:50.000000000","tz":0},"subject":"reconstructor: silence traceback when purging","message":"reconstructor: silence traceback when purging\n\nCatch DiskFileNotExist exceptions when attempting to purge files. The\nfile may have passed its reclaim age since being reverted and will be\ncleaned up when the reconstructor opens it for purging, raising a\nDiskFileNotExist. The exception is OK - the diskfile was about to be\npurged.\n\nChange-Id: I5dfdf5950c6bd7fb130ab557347fbe959270c6e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d409c4bbe542f6b9f3c217aa05ffd8669c815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d409c4bbe542f6b9f3c217aa05ffd8669c815"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
