)]}'
{"id":"openstack%2Fswift~818565","triplet_id":"openstack%2Fswift~master~I1388f1dfd4e0157719d651d9b133784b8a40f032","project":"openstack/swift","branch":"master","topic":"p-ssync-lock-timeout-is-not-error","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-11-23 14:54:13.000000000","reason":"Alistair Coles replied on the change"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2021-12-03 18:40:59.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-03 18:40:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1388f1dfd4e0157719d651d9b133784b8a40f032","subject":"ssync: make receiver lock timeout warning not error","status":"ABANDONED","created":"2021-11-19 15:42:10.000000000","updated":"2021-12-03 18:40:59.000000000","total_comment_count":17,"unresolved_comment_count":9,"has_review_started":true,"meta_rev_id":"59b4b3c1854eb1e10fb354e400abcbc7b667c32e","_number":818565,"virtual_id_number":818565,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-22 12:26:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 17:34:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 19:36:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-11-22 15:52:19.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"c4e25490a7f1843d2a854f634c5100a885ed2150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-19 15:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8736894ac49c6522264bca3428ef2df3b4163a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-19 15:57:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a3f5bb7302ae096c49db9fa05497a92b17342b1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 17:34:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d2fdd3a614464399cde6537d817fa4 : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/e4ae978810e449249d2b9ca6540d003e : SUCCESS in 1h 06m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b57dc4790274130ae6952ce507ce94d : SUCCESS in 49m 26s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e3a224fbf1484e0c86b5fbc34f7d1d98 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3b839f9afa4845cc80eeb7f9fcde6546 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c39fa1e0bf7f42598accd32a16d6af68 : SUCCESS in 18m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/caa597276fcf46b3819167ac768def92 : SUCCESS in 22m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3591f3a6a182450097ebb8b00701b7c6 : SUCCESS in 17m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d25ee3e6b0074759b6c1ee15c9a15828 : SUCCESS in 17m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b31fc40bbb2542dcbd75f4e30b6767d1 : TIMED_OUT in 1h 31m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/911a84099a5f44558408f79b6224cd3c : SUCCESS in 19m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea27eea0479d4ddd86a0de1dfb6eacaf : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/62d8f083a43648e282c253e594641a8c : SUCCESS in 20m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5213ab4621b40348897a491306c3d8f : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e3eca7c773d9469f8c2e3525c172019e : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac5bf3f832194aff95e8f9fcd3ceba15 : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b617a09c75784c8ab075b341f2a25159 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25390871ff7c49e5bc52e526c515b55e : SUCCESS in 21m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39e6eb7b32ae459c96607a4fdc5ef2cd : SUCCESS in 47m 52s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f14441ccdb3f48e3935d654d43383020 : SUCCESS in 51m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/66b96322ef8b4ca8a79a026bde55708a : SUCCESS in 46m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4597fc7a0fa94188b385565ec59db080 : SUCCESS in 42m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6307beb6bda7481eac3170c3b139e60d : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78a5eeea5a814db08b6cd0656bcf974f : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6df9eb0378ad47d88436f4e453be75d6 : SUCCESS in 20m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84437071880c47a7a8751d2fef0b01ec : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":2},{"id":"255df6fd93fda8ce12688eebaf530250df37bfe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 18:25:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/791301de3a694a6bbb2a946a265eca6b : SUCCESS in 38m 24s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5bb8ca4e9a49496289f359612f068372 : SUCCESS in 42m 25s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/321ac806cc0446b583bb97cd718f5987 : FAILURE in 2h 24m 36s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74f915b52317477b8a516de438a1ec2a : SUCCESS in 49m 33s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/27341c9784864f81842d908606f9f118 : SUCCESS in 1h 02m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc5283d4ce1a9a35a92e2ef5a82e9854a18c1d88","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-19 19:36:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2148666f4ec731d3e1f6ac1a1bbec4f157931215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 20:36:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3cc44088794071845dcf7f1e92f64c : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/740de0ab0bfa4b448bd24aea0f9f02ef : SUCCESS in 44m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a49010cf44f474e9ea4564abce7c41e : SUCCESS in 49m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fbf374e85b78467b8a4f0fb512b369b8 : SUCCESS in 7m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba18b74fae2e487b852722fe4d42d0da : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b56493ae5c549f78ac533be9705ee85 : SUCCESS in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f9cf67c7e3d8472c9916c8b8eb50ef1f : SUCCESS in 17m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bfd346521edd40ceaa3986a17fac2575 : SUCCESS in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/098b31c92f984eb2a21f2fe998c50daa : SUCCESS in 23m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a4947e53f564b4d9085831cab18345a : SUCCESS in 22m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae12cd82469d42129104f6b9f2c4add7 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ee165a5250e45aa830b6b7180128d81 : SUCCESS in 35m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a85533ad02ea49ab85dd5e6cd05efa1d : SUCCESS in 19m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/699dadcef2eb4f1dbf43f61f327cf33b : SUCCESS in 12m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1455ace7ae7e400a87d4b32ca5e77f92 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9473969636834afcb5f049f845fda8f9 : SUCCESS in 13m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc6369181aef49608239b3c7b9bf0174 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d412d5f955674aad9fdcd3b5908e76b4 : SUCCESS in 40m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4af8280d9a474ddfa7fccfeb9c59acd3 : SUCCESS in 51m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4337e1a46a824d3395fbf7f4d92178b3 : SUCCESS in 51m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d42464959b4461bbac2fec26e6d7334 : SUCCESS in 45m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f44524843b73446fb51f33e42372251e : SUCCESS in 25m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34384c33467246a8bf4c98f2eeae7fe1 : SUCCESS in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d623dbe93010435c841a3d9cb9e6a3df : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e1bad0ae796487f950ebd6384f033b5 : SUCCESS in 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fdbc1b3374d4005ac0ed1db052f8b23 : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8f3dc7bb9102ce58a09472c7d5eb8e8d02ef6c4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 21:50:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcd6b6faa08c478892447ddd49353eb7 : SUCCESS in 57m 50s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cb548b154c94b0f885445c974a6b941 : SUCCESS in 1h 02m 19s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/e4df75916bf54ed49384f48769040df2 : FAILURE in 2h 10m 27s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/01e746b828a44ff0b699da8fc69e6607 : SUCCESS in 30m 34s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/27bb55fcf0a744e1aaa29fa4ee81ddf6 : SUCCESS in 29m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0cfc0259ae199e6c831ec253cbc55159219632a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-22 11:12:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"92c9454b9d665cf575e9fea03b7793518d9a9b7e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-22 11:13:28.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ed41cc6b1be5809594daccb9a5314382d7f841be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 12:26:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5846a862651344e8bbfc948d76c0ad1b : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/3827ff7e182b4287ad10d79ef687e86f : SUCCESS in 59m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/85639f6ffc3c4d93b00b98df95cc7388 : SUCCESS in 51m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/915ad3d736fe4d70ae7719f2908a8caf : SUCCESS in 8m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e23b61c00ae8496cb79d55e20823f9e3 : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ffd81486ad94802baeaad6235439e20 : SUCCESS in 15m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd8c31ba0344ce98a01624b7a0736a9 : SUCCESS in 15m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/42c4858d9b0e4b9db3f051ece1648470 : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0a25fba1bb6c44b7a59288d5de9b479b : SUCCESS in 15m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a6ce2cfd74414170a687acae07460bce : SUCCESS in 17m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3a362a0fabd1405a88ce96e05e23c07e : SUCCESS in 28m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f3f29c62093430ea20201963315bc52 : SUCCESS in 19m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee29fff1dc4f441e89c26b7e75787951 : SUCCESS in 26m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e753a627c96c4996bb86895d9cd05a39 : SUCCESS in 23m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5c26ed1326f742dca77407a664105b9f : SUCCESS in 15m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cea0670a6dc0484e89113d9fde014104 : SUCCESS in 18m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa2c38aca5fe472f90958a82691ce65f : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b1e8ad801e1d47f7986bffe4275c00cf : SUCCESS in 23m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/474fa1a94b7f478f9d340befe5d21fd3 : SUCCESS in 56m 09s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/85e9a274b6b74ffe912eaa5e81bea8c8 : SUCCESS in 1h 07m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9efc99fb1c2c4af695bbb2ea0dae6a39 : SUCCESS in 40m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/12a08436428f4968b462bb35aba64c4f : SUCCESS in 42m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bee696934eb4a2aa58de964f1acd49f : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2cada2413484fdf9a4ee930232f7f0a : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d430f2c3b2b4512aa3d180e488730de : SUCCESS in 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c313fef19db342518af10129bca26902 : SUCCESS in 52m 20s","accounts_in_message":[],"_revision_number":3},{"id":"5ea94021a69822ff3f7e6ad99aec85cccfdb636b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 12:43:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c09f05b3babd4853a747d9df14138c5f : SUCCESS in 32m 54s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59ec39746e2e4e25837a8ffdf40f60e9 : SUCCESS in 33m 24s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/2a45218a9f4143ee8f931c09a51a92d9 : SUCCESS in 1h 22m 26s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9828c2114eca4d6e96ee208c49576272 : SUCCESS in 38m 11s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/828e385a53e14f308d1e9df1cf4e8fa5 : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"30c90a9343a4311cac96234d40e6141edd59f932","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-22 15:52:19.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8cf9581f05017d297778c6978f744e4f031b7ef5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-23 14:54:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d8ba50335a7cb5d9630e295a2de363b9b3bf6c71","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-23 14:55:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"59b4b3c1854eb1e10fb354e400abcbc7b667c32e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-03 18:40:59.000000000","message":"Abandoned\n\nthe value of this change is questionable given the design/style concerns raised","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5bf21649f061cf528d7aece0d4350eba3f98a32d","revisions":{"fb1669b43d57a3019c22945aa0074fd6a84970b1":{"kind":"REWORK","_number":1,"created":"2021-11-19 15:42:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/818565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/818565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/818565/1"}}},"commit":{"parents":[{"commit":"f91074f376937699dcb0901a9c46030032730773","subject":"Merge \"Make cmp_policy_info agree with the API behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f91074f376937699dcb0901a9c46030032730773"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 15:35:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 15:40:52.000000000","tz":0},"subject":"ssync: make receiver lock timeout warning not error","message":"ssync: make receiver lock timeout warning not error\n\nIf the ssync receiver fails to take the lock on the destination\npartition or device then the sender should exit gracefully, logging a\nwarning, but not flag it as an error. This is an expected event when a\ndestination node is busy.\n\nAlso, since this a graceful teardown, we don\u0027t need the ssync receiver\nto force a disconnect.\n\nChange-Id: I1388f1dfd4e0157719d651d9b133784b8a40f032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb1669b43d57a3019c22945aa0074fd6a84970b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fb1669b43d57a3019c22945aa0074fd6a84970b1"}]},"branch":"refs/heads/master"},"d5000148a63bce569c9c369284f5491ed7bb74b0":{"kind":"REWORK","_number":2,"created":"2021-11-19 15:57:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/818565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/818565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/818565/2"}}},"commit":{"parents":[{"commit":"f91074f376937699dcb0901a9c46030032730773","subject":"Merge \"Make cmp_policy_info agree with the API behaviour\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f91074f376937699dcb0901a9c46030032730773"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 15:35:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 15:56:53.000000000","tz":0},"subject":"ssync: make receiver lock timeout warning not error","message":"ssync: make receiver lock timeout warning not error\n\nIf the ssync receiver fails to take the lock on the destination\npartition or device then the sender should exit gracefully, logging a\nwarning, but not flag it as an error. This is an expected event when a\ndestination node is busy.\n\nOn the receiver, the event is upgraded from debug log level to warning\nlog level.\n\nAlso, since this a graceful teardown, we don\u0027t need the ssync receiver\nto force a disconnect.\n\nChange-Id: I1388f1dfd4e0157719d651d9b133784b8a40f032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5000148a63bce569c9c369284f5491ed7bb74b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5000148a63bce569c9c369284f5491ed7bb74b0"}]},"branch":"refs/heads/master"},"5bf21649f061cf528d7aece0d4350eba3f98a32d":{"kind":"REWORK","_number":3,"created":"2021-11-22 11:12:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/65/818565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/818565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/818565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/818565/3"}}},"commit":{"parents":[{"commit":"5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a","subject":"ssync: restrict max objects per sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5259ada4c4cc6447a0ea1b4b1fb62fe039c4893a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-19 15:35:56.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-11-22 11:07:03.000000000","tz":0},"subject":"ssync: make receiver lock timeout warning not error","message":"ssync: make receiver lock timeout warning not error\n\nIf the ssync receiver fails to take the lock on the destination\npartition or device then the sender should exit gracefully, logging a\nwarning, but not flag it as an error. This is an expected event when a\ndestination node is busy.\n\nOn the receiver, the event is upgraded from debug log level to warning\nlog level.\n\nAlso, since this a graceful teardown, we don\u0027t need the ssync receiver\nto force a disconnect.\n\nChange-Id: I1388f1dfd4e0157719d651d9b133784b8a40f032\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bf21649f061cf528d7aece0d4350eba3f98a32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bf21649f061cf528d7aece0d4350eba3f98a32d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
