)]}'
{"id":"openstack%2Fswift~819199","triplet_id":"openstack%2Fswift~master~Idde58fb9d986aa30fc94eda16e968aa895500174","project":"openstack/swift","branch":"master","topic":"yoga-python-testing","hashtags":[],"change_id":"Idde58fb9d986aa30fc94eda16e968aa895500174","subject":"Updating python testing as per Yoga testing runtime","status":"ABANDONED","created":"2021-11-24 23:43:23.000000000","updated":"2024-03-04 17:47:04.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"39c464a21235a27c377078f65620f05a03da4f8c","_number":819199,"virtual_id_number":819199,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-25 00:50:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-1,"date":"2021-11-26 15:41:43.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-25 00:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"11ee04da6b240e148c4b6d4fdbd7f90a7fde48d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-24 23:43:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"005d0a2d08379ea31ef2d69882708187aebd2470","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-24 23:51:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8759840e0d399741dd2372911fe3ededd5726862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 00:50:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/46805f8f00e244ca8637475144626006 : SUCCESS in 2m 24s\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2458b137fc0c4ef685931ee0a9e60812 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a85805d7b8482e806e4d7d8e405642 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a013094602e245f2b8dce4df46aebd0d : SUCCESS in 5m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0a79054f112e4b00b262103aa5cc37c1 : SUCCESS in 51m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8bbe5fa975643d2b8d667666191192f : SUCCESS in 37m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef90d7e679774535a50b994c407abb06 : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/88f19bf44f0b414b93ea7605c2646029 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/74989e74621248508ea04dd1d7ac7e7e : SUCCESS in 22m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f1068ea16f154080922a389f00b4624e : SUCCESS in 18m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fbe850799a4343e881c6fe999e6aa696 : SUCCESS in 21m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0975f84d8c24b68b7c71acd8973b622 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0493096dbf7a4a90b8c7477255bef256 : SUCCESS in 24m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bdeb0bcffce945a8ba6cd684edb5049f : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e23b396d34a64593ac4d93fafa8a51cd : SUCCESS in 22m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b6968ce5a6f74a1b94099b623c051c8a : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e02e4a5edd284f53b97ee3c4c1b92b8d : SUCCESS in 15m 34s\n- swift-tox-func-py39 https://zuul.opendev.org/t/openstack/build/037493fdedb84e75bdda17ef628359ad : SUCCESS in 22m 57s\n- swift-tox-func-encryption-py39 https://zuul.opendev.org/t/openstack/build/d176f507f1584632aaa705ae6a497204 : SUCCESS in 16m 07s\n- swift-tox-func-ec-py39 https://zuul.opendev.org/t/openstack/build/6a54715b512e4cf2bf5c158c936a8850 : SUCCESS in 24m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2ef3fbb07cf43e1adde03a00ef077ac : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae37a20f1c084cca969db7a7ad57893c : SUCCESS in 27m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d81e24b8ba9e4c74a4f436dc56401dfb : SUCCESS in 54m 48s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/fcf98330cd734a1cb875490dfc99e5a7 : SUCCESS in 59m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/16863d0b67b5442da6d669b58a2bdeaf : SUCCESS in 40m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/48668377bb36424094ac3e097adfc7eb : SUCCESS in 45m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d790004d6e0473aa6fe8a1d2fcd19ed : SUCCESS in 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea603edcfde48798c895a88b0613f76 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3501a18840949f3ba5fed5da20487af : SUCCESS in 16m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a900687924c146e38eaff1751f886006 : SUCCESS in 46m 41s","accounts_in_message":[],"_revision_number":1},{"id":"52c6164777bd8a6fd3401a6d45cb98d8d8395fb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 01:59:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49d211b9c3dc4b6582376cd283ad6dc4 : SUCCESS in 30m 50s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5acc0ac44e77466fb46627641aba89ec : SUCCESS in 49m 14s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/9412ead1437d4651aef4f56fc699429e : SUCCESS in 2h 12m 54s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bee9f2aa2b584158bd1a94d2afff65a9 : SUCCESS in 49m 31s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a91be39473345e0b24c8ff386b73194 : SUCCESS in 33m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a209f874fbf7bc7a573c88e13d667ec6573940e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-26 15:41:43.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39c464a21235a27c377078f65620f05a03da4f8c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-03-04 17:47:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5910ded95d37be5d38c1a94354755107c39d8baa","revisions":{"5910ded95d37be5d38c1a94354755107c39d8baa":{"kind":"REWORK","_number":1,"created":"2021-11-24 23:43:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/819199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/99/819199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/819199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/819199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/99/819199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/99/819199/1"}}},"commit":{"parents":[{"commit":"092d409c4bbe542f6b9f3c217aa05ffd8669c815","subject":"reconstructor: silence traceback when purging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d409c4bbe542f6b9f3c217aa05ffd8669c815"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:37:49.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-24 23:43:22.000000000","tz":-360},"subject":"Updating python testing as per Yoga testing runtime","message":"Updating python testing as per Yoga testing runtime\n\nYoga testing runtime has been updated with py38 and py39\nas voting and removed the py36 testing. Unit tests update are\nhandled by the job template change in openstack-zuul-job\n\n- https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/818609\n\nthis commit makes other required changes in fucntional jobs and\nupdate the metadata in setup.cfg file.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/yoga.html\n\nChange-Id: Idde58fb9d986aa30fc94eda16e968aa895500174\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5910ded95d37be5d38c1a94354755107c39d8baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5910ded95d37be5d38c1a94354755107c39d8baa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
