)]}'
{"id":"openstack%2Fswift~819387","triplet_id":"openstack%2Fswift~master~I5f7bf65b4e747b5a93b79328fac640b791ecdc95","project":"openstack/swift","branch":"master","topic":"expire-mpu","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-02-16 02:13:33.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-02-16 02:13:33.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-02-16 02:13:33.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-01-28 15:41:45.000000000","reason":"Clay Gerrard replied on the change"}},"hashtags":[],"change_id":"I5f7bf65b4e747b5a93b79328fac640b791ecdc95","subject":"Proxy: override user_agent with backend_user_agent","status":"MERGED","created":"2021-11-26 00:04:11.000000000","updated":"2022-02-16 02:19:36.000000000","submitted":"2022-02-16 02:13:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":54,"unresolved_comment_count":23,"has_review_started":true,"submission_id":"819387-expire-mpu","meta_rev_id":"f57dd1f3ca9743d0426f062659631e525f974456","_number":819387,"virtual_id_number":819387,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-16 02:13:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2022-02-14 18:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-16 01:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-16 01:03:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-26 01:28:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 08:51:30.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-11-29 16:44:07.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-01-21 07:27:12.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2022-02-16 01:03:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"7321da3227c24a6a2591781a6c2610447ad993a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-26 00:04:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98145fc470227ed18de80863fc908a36f4f4d130","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-26 00:10:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6731c5bbbe906b47956c5db2cad8909c6fa97c82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 01:28:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e79e21642f1648dca6829f05a9a80d16 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/3f376907e0d148f6b42913f95dd0c79e : SUCCESS in 1h 10m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8601e9a4817f4657adf67f61b7412618 : SUCCESS in 48m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f847c4433790492492c26d2f1f866b0d : SUCCESS in 7m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4ba23d7b02ce47e6bd0e371b89781e96 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b9d809c2a7c4b10a5f59ba1d5b563d6 : SUCCESS in 18m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95c947bc69ea4acba9dd7d5d2f6d1e8f : SUCCESS in 16m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad7634b8a757423d904fdc54eea3ee04 : SUCCESS in 15m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b12105cae94542b69f28feb11e77d035 : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d9e8022635154fb88dd23dd1688547b4 : SUCCESS in 14m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f63a43365c9c41729ff0e316d7c34609 : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e774bf49dec429b945298f860b70a94 : SUCCESS in 17m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f157ba7641b6456ba78ef4d7e5d3ff3b : SUCCESS in 16m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b04ccf48ae34eea9de6113f2e65efb5 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6601df83682e4e0aa9f2a17bad803157 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a1682b86a34049ceb38d4d08bd964c73 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a10715982e148bb93a0c512c46c80e6 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf721f3617124542a9fd6c96e5f9d002 : SUCCESS in 21m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/77270adac2d04b2cb98a3040fad570b1 : SUCCESS in 56m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8bb0b17a824447efbfff9f46c7819df2 : SUCCESS in 57m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e95f6dc0e3bc40ebade9ee1be5b56143 : SUCCESS in 39m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32ea4d75ffc649409f5859f203ed1645 : SUCCESS in 45m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c173197f25db42d782afba900491c1ed : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f8c7d338b0f40dbb76e8a18cb3e6458 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a82878e185a5464d976ad7e69645cbef : SUCCESS in 20m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11cf642d16b447e3b84fefde2422f9f3 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"8fdd1776c185b714ee1a56b68c7ddf8b7f677ef4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 02:07:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38c04265c2bb44e18c1612fa8eb02c0b : SUCCESS in 31m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2943fd755b0f4ee6928f81eb12b45758 : SUCCESS in 29m 58s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/a0d0b2abf79d4c5a8ce8664ee9d9c87b : SUCCESS in 1h 26m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/75d7227632b64478943634e0f71d568d : SUCCESS in 29m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdab6b30cf1e407791c34794d5da2c8f : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":2},{"id":"1496be4653db664cbd536942dc384dd360b43e6f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-29 08:51:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"49cbb3e39ced407095835e422842a5697a888de4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-29 16:44:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b0a325bcb5109dc7bab6924bf87770cb092ea32e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-11-29 17:31:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a1108a823ebadcef4befbae920fe2ac00d5c2d9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-29 22:31:16.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d6924bc29f5be3390c2448e9f231213c137dc03d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-11-29 22:50:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4e8d429247579b5c2ba8d1e2edeeaa181d49838","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-02 04:27:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"92f4ff80020be114bfef1a14260c31e6edeade0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 05:31:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642f32014bfd485781585e74523fb4cb : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/8d5b3c6bb5d840a8a8f44b64d04d5b15 : SUCCESS in 1h 02m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a684df188ad4c7cab0580f0194182a9 : SUCCESS in 44m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ef7466e730ec4484a16b8143a66b18c8 : SUCCESS in 8m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7df8c4f337a04e12853b279928a52845 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5982dead33524ae48ded107639d12d07 : SUCCESS in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90bdbf310d5f451595111d06d41b4adf : SUCCESS in 16m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1fcc2c90152d48a1b88c063b9ce6c0a1 : SUCCESS in 15m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/02d89f836b3646a4916a297c0138b13a : SUCCESS in 15m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d886a18d6a949e880552d6090b84c72 : SUCCESS in 14m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cce90abb28294b5d9da8f541b2a0ddd9 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8eb976437da9403b840eae9fa2c7761e : SUCCESS in 17m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/53ea18a307e04e78b1abc5a48ded5201 : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ed5d1b75b27f439dab2d839f6874e59f : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ac9eb743a63a4001bf33287c9fcfbd94 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1eadb4d8567f4501886bf664104c2f9a : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c6659f3b64e4273b5a0013da69a540f : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a0b802d1b9444f1bbb16aabe01d2689 : SUCCESS in 22m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c1c707f5efb448da816da331977989a : SUCCESS in 50m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6c41fc216513408fb7977d4099c40e03 : SUCCESS in 42m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/005408d993d74d7689c89f84d7db4284 : SUCCESS in 39m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a92578b76658492b9fe26e106f8ac0c7 : SUCCESS in 37m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bc30fb3de8d40e3a85b3d65d00b8c58 : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afbab27dc804c54aa4861d6205e2a26 : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d41b37a937594bd2a7d936d51604c85a : SUCCESS in 23m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e33d926386374cb1a2b5cfca061f8efc : SUCCESS in 45m 29s","accounts_in_message":[],"_revision_number":3},{"id":"74a08e0a3e0023427a86c55455e306d232fc28a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 05:49:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba4f7f39568b4613a1766f424c16f405 : SUCCESS in 53m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/545fef989d824056a640381183efaa5a : SUCCESS in 28m 33s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/118b2714ee144689b59f5ed0363a21b0 : SUCCESS in 1h 15m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cde1d7e984e741ac8e827f79991f73a4 : SUCCESS in 1h 10m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/829d4621fd444f22abe9f044be9be64d : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":3},{"id":"51546f4aa510a302654a0fdc60b4a9076533876f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-10 23:15:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b66b8fd737da4bef0583adc22fb5f115a9362496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 00:48:51.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc79811be41a4fbca5f08e5a336119c0 : SUCCESS in 34m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a6a8e9060c24cc082c9b239a4131479 : TIMED_OUT in 1h 31m 51s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/53cdac920775449aa096e53bac640e0d : SUCCESS in 1h 27m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dc8d3bf52c1487ca520194f4d755602 : SUCCESS in 36m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0deadf02bbf849a4a65380175c8f0600 : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":4},{"id":"ca84e5275d160722ac8842e00ee580fe1fe46d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 00:54:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23f67d77ae241b1a74a35637c326d0b : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/c25b3510dc2740b49e0a0e65952f17b1 : SUCCESS in 59m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/00c8209d1d46495c8f58b595bb6482e4 : SUCCESS in 50m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/081c7ae5e0c348408fdb4c287f770983 : SUCCESS in 46m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9fd1aaa9629f4f649a4c7b692b923a26 : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1cfe7896fa564ff1b22a4bd44424630d : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/839066afe1df412fbe58b600e60bffb7 : SUCCESS in 16m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/085b063477fc4fbfbd70c9677326ed73 : SUCCESS in 15m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b2b756b27a894cefa38f3c8385cc7a88 : TIMED_OUT in 1h 31m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/84c3c56ca58d44948ed0e7c216361210 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/83e53bbfd2f74719a433e269b344590c : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc5a7b94c7ba403093e3dce5af3ee6af : SUCCESS in 18m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/802d3b28133b48a3a5488ddeecaa6ea0 : SUCCESS in 20m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/db05fe6fb7064d6d8918e2decd7ff66c : SUCCESS in 22m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f627bd33494b4ace920fb4cae9c42bbb : SUCCESS in 23m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/878012d8ac3c4766b11bd5a4f6f3546d : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18fa1495e9b842e2ae59a69365eec3c5 : SUCCESS in 27m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/03bea46378834c52837bd6aafea8d5f7 : SUCCESS in 1h 01m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fb70250891142238d21577a498d1134 : SUCCESS in 57m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00f041efc7ca456a903ea009d058ee71 : SUCCESS in 50m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00e2847411bf4e318530a4a9b46dee5d : SUCCESS in 42m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b517ab7315473f911872636188cd9e : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e03a5099b63344b2b408cca4295bf8aa : SUCCESS in 7m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/722354ad09f54c66bc4981aa4ecf7d60 : SUCCESS in 29m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d6bc7f540fa4867bca07f9f7ac37e90 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":4},{"id":"4847ee510e475771fc781a76a7acd99ff9192ad8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-11 01:21:31.000000000","message":"Patch Set 4:\n\n(11 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f892f3b97a57f8ef2001879b7b695970541d1d63","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-12 06:00:30.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"826874dd8436cc7aee34494600800aa8d1a58da4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-20 02:27:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4525e7fd4d5a35262a2ee7a364050aef26115947","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-20 02:55:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c9522c724b433809683e46fa4a9df768eb7c34f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 04:34:55.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8829ef6149134638aafa78f3a60567ed : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/234d94958b5b43f69f800f32a1047dd2 : SUCCESS in 43m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69d5f37cbfff4b1195693f3cddea7853 : SUCCESS in 32m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82ce1f869b3f4c759f8dac396a643f87 : SUCCESS in 36m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/712badb4f8274b77b7e85785f9072ce3 : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bddb4a03644c482084414d0ca6b3933e : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/33ce70ade3a94e599e1e5ca06ba52f86 : FAILURE in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/532904511ec24c21b0c3b942ddd2236c : FAILURE in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/293461bfe20f4327a30caa442bf93ac1 : TIMED_OUT in 1h 32m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07281ba17a4045529c45101af674576a : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/903552fd65d541a6a93de666fb25be0f : SUCCESS in 19m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/42f7958274ff4b89810a75197ff3cc43 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d8a7505cf685477790ea9a46b7715f6b : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68df3a96d8494cd6ba4d5f0ff39c147f : SUCCESS in 17m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/608aae6a114246e4a21056e6ae8ad9aa : SUCCESS in 29m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab5a78a8dcd64551848480bafedfd86c : SUCCESS in 11m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c7f17b19c3347578bf1d4ea8df92e61 : SUCCESS in 21m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/62dcf4474a224a5685230e922736adc6 : SUCCESS in 1h 02m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbe76022490e4c7f9c032af9863bd896 : SUCCESS in 1h 06m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/780c6fb6a82a45379bd7f5d0e89f1b83 : SUCCESS in 42m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34b792e4a139418bbffc873a41bcd203 : SUCCESS in 47m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69fd0aabe0804d9faffc9741ac5c742a : FAILURE in 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e528e693abc49dea8973c891d28726d : FAILURE in 7m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d04ed6ad60644deab3c4ae9cb6eab125 : SUCCESS in 19m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c5a2a529f674df690b058953f47f945 : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":6},{"id":"947e599f99790985252192864d766a8de4de856e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-20 06:08:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d7452a12146cffe6e2f5863d454120b3976c3904","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 07:13:48.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cadcc49e5e4f51a5a05e5700f7ae9a : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/aad1d7443e66459ba8860695ad2dd6a5 : SUCCESS in 54m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ef14c1df384f4e069c7663c6b3334f54 : SUCCESS in 46m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fda97d40ad8248f7bdd2fa23a97aca65 : SUCCESS in 53m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/111d6483f276439bb4168ac90fbfd503 : SUCCESS in 9m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/acc3b56e2196455194373c665eda21f4 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c32b20444ed64a288072ed50edc34440 : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cac8889c36b14fb1987e6850261dacaa : SUCCESS in 15m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e2ffe6fc8a6495d865dba409fd44112 : SUCCESS in 20m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd860c93e22041da82285b062066b0ae : SUCCESS in 21m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94573fc0db6d459fb430d682280b09e6 : SUCCESS in 20m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/767c540ec91d4df0b6bc8ab14f1a2fac : SUCCESS in 21m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/337b580a8498452792119cf713fb17e6 : SUCCESS in 20m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4eb93708daae4f3a9ebc98deecc9adeb : SUCCESS in 19m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ab4e3b18e4994558bd49423cd8a4f70e : SUCCESS in 20m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f53ca3a1cfdc4c2397cbda42cb1458f3 : SUCCESS in 5m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e8b6e443abd415f9606a84ad4adbd74 : SUCCESS in 24m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dca77955efd04885811956fbec518364 : SUCCESS in 42m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a601c78001a40f0bb8a278945d89405 : SUCCESS in 1h 02m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/115fe34756a747868322e7618f3f5c5e : SUCCESS in 24m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b80129067da14b98af5b3ea1a97e5917 : SUCCESS in 44m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f68e03ec11b4ae9914bce2c74c932e2 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8edd2ec46c054c3d9c743dc40769f549 : FAILURE in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/229c13d8d7134763b09c401fc9f15fe5 : SUCCESS in 16m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8784b8ea7080408ea8381c2a05bc854f : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":7},{"id":"eae3ba31d5d4d27710b87102d65c777dd1268ba6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 09:18:28.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea4fea0cf1eb44068414f9abc8868b3b : SUCCESS in 31m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd66b9db435f45e69d703893cbbefead : SUCCESS in 33m 43s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/e129a2e37636462188a3bacb9826db93 : TIMED_OUT in 3h 02m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2403845d96b64a10ab187f48758f79ff : SUCCESS in 35m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8382242e17949dfa68808fb5cf82498 : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":7},{"id":"0d8fc5db28bcdc2998070d468dd41f94f8dd86af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-20 21:17:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f4aaa008b6d7f1b10576b697a10f863300869086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 22:28:58.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4288075aa29749799ba108fcc8e73f12 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/5c9de884d230430cb5d95f173457fbdb : SUCCESS in 49m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/107a6d6cb33f48909aa596c4b0a7d1c3 : SUCCESS in 35m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7aa5f3eea38e45d3a58b80cc1f9eaee7 : SUCCESS in 41m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e1543f257f4458686070b9b09fa2e06 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/65dde80963f3405e9162d166a95faab8 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c7641869ea24a81b87da63e9e0d942b : SUCCESS in 12m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8309321235e437fa70f1c8a200908fb : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7155ce44f0224758a461ec077223a4c8 : SUCCESS in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/83a6d4edf56243b2b1d00362caf9d72c : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2060da9ff3d7407382b61645c2544959 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ceefd521a0449319f32ddd082f32840 : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b04678c4767b43d89ee896594d09de95 : SUCCESS in 23m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4ddec95e8a484cd5ac155f639ab2dbec : SUCCESS in 16m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8e3be23c45144b14933709206589145d : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/61c98033a1d74f1f8da0852eef12bdc6 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6981aaf05694e50873acfb5f9258230 : SUCCESS in 20m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51557cb1da554c019212698ab05ffb56 : SUCCESS in 57m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c17e079f6e5e4fcbb7c394239c1c6178 : SUCCESS in 1h 04m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06d45891397a494a9b611e337cf6038f : SUCCESS in 39m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5719651f18b14f6a8d6fc52d7ecee1ee : SUCCESS in 45m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0751cf5102b4448aadfca84a8367820e : SUCCESS in 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cadb8c5668e94f229baed6e5602b3bec : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1becc6b3e9b145369b520937272d22b0 : SUCCESS in 34m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03e0642f107c4e858470d72fbf59e275 : POST_FAILURE in 57m 46s","accounts_in_message":[],"_revision_number":8},{"id":"ecba7097df32a8a38d6b77dc437a89b2e71c04e2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-21 01:21:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"98e14487750a307fe54263c33e287ef271e18981","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 02:06:52.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3767d191b79e44ffbfbb80acd48f04ca : SUCCESS in 31m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf9f981989394da5bbfaa41dcfed549a : SUCCESS in 29m 51s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/f4f9fb5c3c5342a68b7a933e59bd0239 : TIMED_OUT in 3h 01m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/446c40fd30584fe88115da07a8671eda : SUCCESS in 32m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/63ad73bae55f42e290811fd5bd882d1d : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":8},{"id":"d89eb4a47b9789188c9b40bd7a8d89321b0e5557","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-01-21 07:27:12.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a486540eedc0ad33f9936ddb309f29577713a079","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-21 13:27:14.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c33384dac1082f79f71ae83562e44b8e45717785","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-01-21 16:20:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"de2c6c17e40a33c5b43593d17b6a30f1e9d56b16","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-24 15:58:57.000000000","message":"Patch Set 8: Code-Review+1\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"29b59bf7e88d7a785b37fe13a028e1afc0e34d2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-27 18:06:27.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"8153027ea144b151a73e6f4ab71824ad6fd145be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 21:31:58.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ccc576e87a24015b1c5a1ed15453ad5 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6fee6906d7be4e67919ff6aa442e0f1e : SUCCESS in 59m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/126dce7f644e443a8cb33ac9209d22ee : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f25cd6cf3c34657b111d30825b1a20d : SUCCESS in 51m 18s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/36159b12636443bd80e654b175168248 : SUCCESS in 16m 59s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b9675f5cb92c49a5a9ce6945e0d60625 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/31f0bbabe3964819b8a8c7715cf18a5e : SUCCESS in 20m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d0eaa2bca70d4e5b89a193cfd29b2810 : SUCCESS in 11m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ecf3f6f8eb8041808666321c16e034fa : SUCCESS in 7m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7b0c8d2fdc504a9c8b97cb24d1f41697 : FAILURE in 22m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a768d0b4934470b91556577b5634d32 : FAILURE in 17m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8d032c9bd1e94fd6888cf092e05ac5e5 : FAILURE in 18m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c623131bf0b442558d0636cc10ad5579 : SUCCESS in 19m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c57cd3ad07b94b4aa130b4eddb0bcab2 : SUCCESS in 22m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f261611630a24dd0a96c2e5c3611f31e : SUCCESS in 23m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1618e6f8bcf943d4b5c80b5b2facd45c : SUCCESS in 20m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3f85303df90d4c61b2de31263bdefd18 : SUCCESS in 22m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed9a373692ae434ab0f781248428707e : SUCCESS in 23m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/600e4d2276e74ef998463cfb9d8ec367 : SUCCESS in 8m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/026f76de6d4d4d5d8b5ec7e87f195344 : SUCCESS in 19m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/156aadc303d74e27a487a0922b7c5fb6 : SUCCESS in 54m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/89f9689c18cc4282a929033728cd254f : SUCCESS in 1h 11m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7867ea19ecc549afaa1c292a14347ec1 : SUCCESS in 40m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f882d098034b4a3b91552637da11b055 : SUCCESS in 48m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cf3eeca20c945498584d20823a16da5 : FAILURE in 18m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c87f63351f174879b137e86232a24ac4 : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b39d05086f8469da1408d00d062c04e : FAILURE in 23m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4698c1b1e8454246b18e88ed5b665552 : SUCCESS in 49m 23s","accounts_in_message":[],"_revision_number":9},{"id":"27f27c7fb8f53f43bc4f7471f673829f28614363","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 01:00:17.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2790dfe8a9ba45b5964a94121a5e4f97 : FAILURE in 38m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ab6b46aacc34451960554f2ca710fd5 : FAILURE in 1h 15m 27s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d3294d4c5f6d4aa994a268dacc50b5d4 : TIMED_OUT in 3h 01m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/153b5fc20ed1416589fac940c86f2dae : SUCCESS in 46m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fd267e2e6974687b069fa5d8033e20d : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":9},{"id":"1b43698d54d1acefbd0aedc3dbb58145902d8965","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-28 03:36:21.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b9398ad167dc04fb5ee6d144dbf0b26b4b13a700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-28 03:59:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"180ed91b2129b937ab6f4f6c7b38300720551e06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-01-28 04:00:47.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"bcf00fc5c08bc21fd7dc60dbd6f420be3dbaecdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 04:53:23.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ebc405bd214b1e8e55229602a766d0 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/a39feb3b27cc495c82431a71b7b8bfc0 : SUCCESS in 44m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/266760a56a7446d09323f43f3442d7ec : SUCCESS in 35m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fe8cd74945b473b9822ff9aa58692d4 : SUCCESS in 36m 53s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c91a235d718547f5928943b950ef1b56 : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/efacea9f4e83402494bedf4e213ee5dc : SUCCESS in 29m 57s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6962ebf0be6b436f97cb473a0afacc32 : SUCCESS in 22m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33c55b19a6874a3db3cef188db803ab7 : SUCCESS in 8m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c9f1735961d4e848f19f505b680d0ff : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c36ddfbc0adc46a58c84fe8c187d2e3e : SUCCESS in 12m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64dc6ca1bde34fe2a096dd5cd47cfdb6 : SUCCESS in 18m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/810d03cdebc541c690016817c69c7eb9 : SUCCESS in 12m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18703fce75514a54ae43603a821da85b : SUCCESS in 20m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0500e0d2f83d424480365163ab3f4f24 : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0817af17c0646c19cb2bc202298ec60 : SUCCESS in 16m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3aae7f41e7564f01aad3d6ec28f8b81d : SUCCESS in 18m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9c480c1b59d645a69b8ccdc34d9de1d9 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c70161fa1ca7406e8a00e450adbf752f : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/534308179c5644bf8dd882f36a7adfd2 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69dadd4ae38b4c6fa2becfab0da49b66 : SUCCESS in 25m 19s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9dbe5a97d59a496f832859775962d67a : SUCCESS in 36m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcbecdd87b8a4d6288a95984c8ab1b00 : SUCCESS in 50m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e123d6c516074fe08e713943e23d2c7f : SUCCESS in 32m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4137163a29f044199e75a6ca2e39b25c : SUCCESS in 30m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fb2fffc24f8433baa0a6fa9bbf66e68 : SUCCESS in 22m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97bcbe971ae4d549461ea427ed6da0b : FAILURE in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4263fb2a96f143599aa3878577e15daf : SUCCESS in 16m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe05b080322d44aaa30b04869b001a3b : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":11},{"id":"a66543ff739022960b4329fa763491f100d0d4dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 06:47:02.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8a15fbf995845a2b10d1cae79bbe7d5 : SUCCESS in 30m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a22b1a0d2fdd4de6b0aae180ebefae15 : SUCCESS in 29m 55s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/892545fcc4314403ba6115d6d5827f19 : FAILURE in 2h 42m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c964a85014d6473fbba4b2ac8d440c42 : SUCCESS in 29m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e89ef395d4e44438bf36dafe744a9de4 : SUCCESS in 28m 07s","accounts_in_message":[],"_revision_number":11},{"id":"efd21e14763fa51ae69065f5a2e4f765f4a3b9eb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-28 15:41:45.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"429aff215aa5bb1040e7e1089ab9d60c0919b6bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-28 15:57:22.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d93f9deef9400824f886fe804a6590a339df1b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 18:03:27.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427e1b9a30ea49078119ecda819e66e7 : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/0c3ae442b9c14845a5e3b673224a9b2b : SUCCESS in 53m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1aac1ef4f2248b7be07adf195d0bac6 : SUCCESS in 43m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/843c2181cd534b76ad6bc356a6fbd234 : SUCCESS in 40m 43s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/20ad6ff7bafe4e5bafd709caef784c91 : SUCCESS in 30m 26s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cc408c0304764ac7b5fc396800f32b08 : SUCCESS in 28m 15s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/15bca2d40b324360ace8259c6dc22d01 : SUCCESS in 27m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7fa3c24489104bf8a365457423e57502 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6d821a8966444f585b0d00d719ffa4d : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4d3841297d44ef183d94f2a9e4a226d : SUCCESS in 19m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c36b14c021d4fc68e9c47500fac728c : SUCCESS in 20m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ad813ce5342c4affa08aba6da7cf6d59 : SUCCESS in 20m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc4c970d62544c41adb6c421e6a1878a : SUCCESS in 22m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/24da6011e616400e9f793f3ab5b5e711 : SUCCESS in 20m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb28422cc9b243b9978ed36e10788be3 : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7d8fb5a47dc84df89bd3db9ab1b390de : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3946e1724f614f609df2f8324f2c16ed : SUCCESS in 19m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7f4de2ff515243e9948e25c5d5cff26d : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8681e961f46e4c7cb7317bf48b791f9e : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78c176c6956147a49278f30cd16f9442 : SUCCESS in 19m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b2e58e54f6548d0a5109342d043be04 : SUCCESS in 56m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/33092181843543c09c2961cd7ee16c24 : SUCCESS in 53m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2fe9e9a15e17454eae3b81eaa5a57e6a : SUCCESS in 49m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17d92d28bfe34a07b35e2fc7a70ce16a : SUCCESS in 40m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22658949440f4f4aa8d900df11276b15 : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4fe198b5f0b4d0b8e4ad4ee6b6ac42f : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e481f20864d43e3bd91b43fc6c7c308 : SUCCESS in 24m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4b2718dbca24b0e9a40614522dd8f5b : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":12},{"id":"eacc61701ce69a40820bfc44f9f9519098570dc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 19:08:37.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce4ab3e5f52f4722bd1be04e7a5a008f : SUCCESS in 36m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ed9088488c54509ae02549459451504 : SUCCESS in 36m 32s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/f03f97729ee448c5a931729ad784dac3 : TIMED_OUT in 3h 03m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd10f2fe56c4415db71e158802135afa : SUCCESS in 1h 00m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0490e84ed2a1444796af3481b021e652 : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":12},{"id":"9438c9291fa2b85c9061ff76276a774a5a73667e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-02-14 18:27:35.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"9e9ba75eac10158ad8b4694ba457bb49bcaf9536","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-02-16 01:03:57.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"17ad826bc042abb56cc14d9130149a882bf014e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 01:04:45.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"8f9b799fbf911be144797b66476ea9fa2014d31a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 02:13:32.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd8e0969acb4ad0acf1270792a046ca : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/10f7810b5dec4023afc9ed2a0fb114be : SUCCESS in 1h 02m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ff2e20c8a1d49bda3735e2508689ce1 : SUCCESS in 39m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95d9959879aa41ee997ab7b6f5b69282 : SUCCESS in 51m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/980034ef304245e7ab1cbad715788f9e : SUCCESS in 7m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/483a1f27c4434fd6914c27a1c2b61a9b : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46a77173d1b84afebc90fd9e16b6a7fb : SUCCESS in 14m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6598e9b5a1e0461e83b3c22f8084fa77 : SUCCESS in 17m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cd0f15af9a4348c096cef8b2bb5ad46d : SUCCESS in 14m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8f6054738b54d61882d5904d28afa29 : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/576753d18571484e9dab222b835388e8 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e594b14765574a12ac3c28d8bde64abf : SUCCESS in 33m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f41bf63f57014e10b3b1ef58c30e2686 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec34c2d8c01f44dbb1e276cf39cfd972 : SUCCESS in 30m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1afcd719872942a7ac2893f89e26cc52 : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4020993a29264954a0a5b151443e079f : SUCCESS in 5m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e0dfdc304794ce789acf839a515fd25 : SUCCESS in 53m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6549f218ba0c47c9b88ce25dc0f87ddf : SUCCESS in 1h 04m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef85bdefae054e909b39725103d6c9a6 : SUCCESS in 33m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2cf47eb355f4fea96e389ebed2aa17e : SUCCESS in 45m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33b6fa84a7034ee98cf520bdf4531b9f : SUCCESS in 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0bb6a8d7fbe49d88bc96081d508c471 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4cbdb8678d31429ca5a8bc7ccfc9acac : SUCCESS in 22m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f384138c9ac84bca88265477c84c8878 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":12},{"id":"e0a46d6829a27d98ccd42d9cdc4f93e2f0ee3bfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 02:13:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"f57dd1f3ca9743d0426f062659631e525f974456","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 02:19:36.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517c450eb6804bf497448f96e4c4608f : SUCCESS in 1m 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c1a1edbfef45459e90861847d9594afe : SUCCESS in 2m 07s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cc11790472634a0a88d478ed8da00bb5 : SUCCESS in 2m 03s (non-voting)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"b0245f4eb5e625a0528e8ce607b1d604201bc6da","revisions":{"82a5973400551ab69e9ef354f1b804ea5becae97":{"kind":"REWORK","_number":1,"created":"2021-11-26 00:04:11.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/1"}}},"commit":{"parents":[{"commit":"092d409c4bbe542f6b9f3c217aa05ffd8669c815","subject":"reconstructor: silence traceback when purging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d409c4bbe542f6b9f3c217aa05ffd8669c815"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request. Additionally a\nset_backend_user_agent is created that is then used by the internal\nclient via a `pipeline_property`, so it call call it down the WSGI\npipeline.\n\nAs to not make any change to the InternalClient API, the self.user_agent\nhas been turned into a property so it behaves as normal, but it can call\nself.set_backend_user_agent behind the scenes to set everything up.\n\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82a5973400551ab69e9ef354f1b804ea5becae97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82a5973400551ab69e9ef354f1b804ea5becae97"}]},"branch":"refs/heads/master"},"23bb05ad6277c4564e9df48b1d78a5bf5d7d50e0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-26 00:10:21.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/2"}}},"commit":{"parents":[{"commit":"092d409c4bbe542f6b9f3c217aa05ffd8669c815","subject":"reconstructor: silence traceback when purging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d409c4bbe542f6b9f3c217aa05ffd8669c815"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-26 00:09:10.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request. Additionally a\nset_backend_user_agent is created that is then used by the internal\nclient via a `pipeline_property`, so it call call it down the WSGI\npipeline.\n\nAs to not make any change to the InternalClient API, the self.user_agent\nhas been turned into a property so it behaves as normal, but it can call\nself.set_backend_user_agent behind the scenes to set everything up.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23bb05ad6277c4564e9df48b1d78a5bf5d7d50e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23bb05ad6277c4564e9df48b1d78a5bf5d7d50e0"}]},"branch":"refs/heads/master"},"d2aebdc77f44adc5dfca83e60603d1db5e01331c":{"kind":"REWORK","_number":3,"created":"2021-12-02 04:27:56.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/3"}}},"commit":{"parents":[{"commit":"092d409c4bbe542f6b9f3c217aa05ffd8669c815","subject":"reconstructor: silence traceback when purging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/092d409c4bbe542f6b9f3c217aa05ffd8669c815"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-02 04:27:44.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request. Additionally a\nset_backend_user_agent is created that is then used by the internal\nclient via a `pipeline_property`, so it call call it down the WSGI\npipeline.\n\nAs to not make any change to the InternalClient API, the self.user_agent\nhas been turned into a property so it behaves as normal, but it can call\nself.set_backend_user_agent behind the scenes to set everything up.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2aebdc77f44adc5dfca83e60603d1db5e01331c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2aebdc77f44adc5dfca83e60603d1db5e01331c"}]},"branch":"refs/heads/master"},"5d4ac2760844a42e3a9896f9f44bdebfb3622067":{"kind":"REWORK","_number":4,"created":"2022-01-10 23:15:53.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/4"}}},"commit":{"parents":[{"commit":"079abc5f71d8932ccd3b10f2c9610597c734f433","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/079abc5f71d8932ccd3b10f2c9610597c734f433"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-10 23:12:33.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nAs to not make any change to the InternalClient API, the self.user_agent\nhas been turned into a property so it behaves as normal, but actually\nreturns and sets the `self.app._pipeline_final_app.backend_user_agent`.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d4ac2760844a42e3a9896f9f44bdebfb3622067"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d4ac2760844a42e3a9896f9f44bdebfb3622067"}]},"branch":"refs/heads/master"},"f2d61e8497ab3c9a6fd663bc0a129b2bfca19233":{"kind":"REWORK","_number":5,"created":"2022-01-20 02:27:34.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/5"}}},"commit":{"parents":[{"commit":"55295005fd13a35cc78bba0adcfc85c4daf45d84","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55295005fd13a35cc78bba0adcfc85c4daf45d84"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-20 02:27:22.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nAs to not make any change to the InternalClient API, the self.user_agent\nhas been turned into a property so it behaves as normal, but actually\nreturns and sets the `self.app._pipeline_final_app.backend_user_agent`.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2d61e8497ab3c9a6fd663bc0a129b2bfca19233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2d61e8497ab3c9a6fd663bc0a129b2bfca19233"}]},"branch":"refs/heads/master"},"3e18773616e0dd7a093b49f564610698ea118e0c":{"kind":"REWORK","_number":6,"created":"2022-01-20 02:55:34.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/6"}}},"commit":{"parents":[{"commit":"26c0f31379ba4b47f84fe13c1f7f09a47abdb15a","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c0f31379ba4b47f84fe13c1f7f09a47abdb15a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-20 02:53:09.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nAs to not make any change to the InternalClient API, the self.user_agent\nhas been turned into a property so it behaves as normal, but actually\nreturns and sets the `self.app._pipeline_final_app.backend_user_agent`.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e18773616e0dd7a093b49f564610698ea118e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e18773616e0dd7a093b49f564610698ea118e0c"}]},"branch":"refs/heads/master"},"23e5e1d19a26bcadc35f18c00c167edc890ea542":{"kind":"REWORK","_number":7,"created":"2022-01-20 06:08:32.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/7"}}},"commit":{"parents":[{"commit":"26c0f31379ba4b47f84fe13c1f7f09a47abdb15a","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c0f31379ba4b47f84fe13c1f7f09a47abdb15a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-20 06:07:52.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23e5e1d19a26bcadc35f18c00c167edc890ea542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/23e5e1d19a26bcadc35f18c00c167edc890ea542"}]},"branch":"refs/heads/master"},"d8d0a414a0be3a63cdb8e870a2f69d741b95d235":{"kind":"REWORK","_number":8,"created":"2022-01-20 21:17:43.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/8"}}},"commit":{"parents":[{"commit":"26c0f31379ba4b47f84fe13c1f7f09a47abdb15a","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26c0f31379ba4b47f84fe13c1f7f09a47abdb15a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-20 21:17:26.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d0a414a0be3a63cdb8e870a2f69d741b95d235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8d0a414a0be3a63cdb8e870a2f69d741b95d235"}]},"branch":"refs/heads/master"},"6fcab124d5382c9f89ef997914f082761cccf7de":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-01-27 18:06:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/819387/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/9"}}},"commit":{"parents":[{"commit":"4f45bd0d502f4e68db03969d60b8fd673443adfb","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f45bd0d502f4e68db03969d60b8fd673443adfb"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-01-27 18:06:04.000000000","tz":-480},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fcab124d5382c9f89ef997914f082761cccf7de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fcab124d5382c9f89ef997914f082761cccf7de"}]},"branch":"refs/heads/master"},"9a26c07877f2e1526751625b224a4f571360de22":{"kind":"REWORK","_number":10,"created":"2022-01-28 03:59:30.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/10"}}},"commit":{"parents":[{"commit":"4f45bd0d502f4e68db03969d60b8fd673443adfb","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f45bd0d502f4e68db03969d60b8fd673443adfb"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-28 03:37:17.000000000","tz":660},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a26c07877f2e1526751625b224a4f571360de22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a26c07877f2e1526751625b224a4f571360de22"}]},"branch":"refs/heads/master"},"bebf46c7115027fcba8be07c68bebc0409093d2c":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-01-28 04:00:47.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/87/819387/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/11"}}},"commit":{"parents":[{"commit":"9bc1c008a56fbc4a69aafd3d3177d7be368b181d","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bc1c008a56fbc4a69aafd3d3177d7be368b181d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-01-28 04:00:47.000000000","tz":0},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bebf46c7115027fcba8be07c68bebc0409093d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bebf46c7115027fcba8be07c68bebc0409093d2c"}]},"branch":"refs/heads/master","description":"Rebase"},"b0245f4eb5e625a0528e8ce607b1d604201bc6da":{"kind":"REWORK","_number":12,"created":"2022-01-28 15:57:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/87/819387/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/819387/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/819387/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/819387/12"}}},"commit":{"parents":[{"commit":"9bc1c008a56fbc4a69aafd3d3177d7be368b181d","subject":"Get rid of pipeline_property","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bc1c008a56fbc4a69aafd3d3177d7be368b181d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-11-25 23:54:44.000000000","tz":660},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-28 15:46:56.000000000","tz":-360},"subject":"Proxy: override user_agent with backend_user_agent","message":"Proxy: override user_agent with backend_user_agent\n\nThe proxy currently _ALWAYS_ overrides any user_agent in requests with\n`proxy-server \u003cpid\u003e`. This is what we want for any normal request. But\nwhen using the internal client, which itself uses it\u0027s own proxy server,\nwe\u0027d rather use the internal client\u0027s defined user_agent, so we can\ntrack the source of the request in logs.\n\nThis patch adds a backend_user_agent to the proxy, that when set a proxy\nwill use this as a user_agent in a request.\n\nEvery member of the pipeline has a `_pipeline_final_app` to point to the\napp (proxy), so uses this to directly set the proxies\nbackend_user_agent.\n\nCloses-Bug: #1951304\nChange-Id: I5f7bf65b4e747b5a93b79328fac640b791ecdc95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0245f4eb5e625a0528e8ce607b1d604201bc6da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0245f4eb5e625a0528e8ce607b1d604201bc6da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
