)]}'
{"id":"openstack%2Fswift~819742","triplet_id":"openstack%2Fswift~master~I1c415ec6fcfe426f71267c2fb7d714c2d84e0014","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-11-30 17:29:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1c415ec6fcfe426f71267c2fb7d714c2d84e0014","subject":"account: Remove logging translations","status":"MERGED","created":"2021-11-30 01:19:37.000000000","updated":"2021-11-30 17:42:15.000000000","submitted":"2021-11-30 17:29:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819742","meta_rev_id":"5e0aac4fcc285504d8e7f42587c0e67fc853003f","_number":819742,"virtual_id_number":819742,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-30 17:29:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-11-30 14:40:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-11-30 14:40:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 02:27:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-30 14:40:35.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"89d4c1e5dd71316b364f741818781ecf16c06441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-11-30 01:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed2688d1959d84d57215e54755630a9d5837e65d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 02:27:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c2191b91fc4dde8204fb3b0b152df1 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/f60c023eac96472ba46f99bac08f0c74 : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/286702a0422f4235be01348820bcfb14 : SUCCESS in 47m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/68124f3d99ae4dfd87041b58b079de4f : SUCCESS in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6e6314897111453ab1c4f334495249d0 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/48f233414a3146b19166e7d53f3cd9f6 : SUCCESS in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf7dac1aad834f18bca3c9e08427199f : SUCCESS in 17m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84c2d7da350f4c928ce3cdf156fecbf6 : SUCCESS in 16m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b41b6923472d46a6b5be03f0c208e455 : SUCCESS in 15m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a73235cdf4954d11b5d3d87497a4ea45 : SUCCESS in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8adf45694baa4654ae0572d1468b6096 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5cb0d130ccca4e36b469665a286d9d2a : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25822bd2639f431ebe1599b19b7dee2d : SUCCESS in 19m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ab396e35a54b46b2972f8f1d745d430a : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8ddfa84ebc1b41aab2fcbd0273639d9f : SUCCESS in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6db23ebeb70443999bd2d66167edf0ff : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/979b92c73bea417d906ca0dba0c534a6 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/557c28b7978247baa0245b898b03e1b9 : SUCCESS in 25m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/faa94873059b4660b4b5a34c592ab684 : SUCCESS in 51m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c7294ddd3fb141fc87434bb18054bf2f : SUCCESS in 54m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b168020d150c42eabe192f73a567e80e : SUCCESS in 40m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6b5f8ec9e73d4b9988f572c731cf51d6 : SUCCESS in 45m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c664fbaa0054a919ede238c10091368 : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1da7c8f9be19478588f7d6c8b292d51c : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94873696238c464eb1bd699c69b9b6fe : SUCCESS in 21m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a23287723dfd46938593423c5c01f13e : SUCCESS in 44m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d9aaf071b655f0d9f05bff02739b0af03db500a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 03:54:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd7c40addea3474d9872f249f3395cc5 : SUCCESS in 52m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ac15a19c5e046fe92aaaa5c45c15587 : TIMED_OUT in 1h 32m 18s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/4e9038dc4ce449e8a793bfcef0e5b6ab : FAILURE in 2h 32m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74299c6eec9846b4bde1b59960bfcf28 : SUCCESS in 1h 01m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f56b3a4978e04208acefce02145cf0c8 : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9d7ad6df52286fce14d77ace22498dffdcd0f22b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-11-30 14:40:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a0e80304fe9d5a37ca54ddc28c5e55f58f9a8ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 14:41:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2c890d2be83b35983f65c004b36fa693f2a8bce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:29:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f97ebcd90e41309fa4e89d8079cfc8 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/019a0e5caa7f4d30b29e8de752c6027a : SUCCESS in 53m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3cb6e892c5c346e6981c78b97ecf8728 : SUCCESS in 41m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e4fc4f9c68db4865a8578cb8953943cb : SUCCESS in 8m 32s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/10c31376eaca42bb969cc43e797acbc9 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a10e3d6931ed431b816a65aaa865f683 : SUCCESS in 22m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b735808f873e43a0ba0bec13480226a3 : SUCCESS in 18m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/554fd17b9bfa41b38312d719a76740d1 : SUCCESS in 19m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/163d0028ceb54acb893c77a6d305aac6 : SUCCESS in 19m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/511d3fb4be374c74b72efa196c50a323 : SUCCESS in 20m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19cb3a22cad84373b305a4beeef37fc7 : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ae94f4148e14f4cb8458934cfd863d4 : SUCCESS in 24m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c827a49bc9a4b93b7cdf4590e72bb0f : SUCCESS in 22m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d63121ff397149ecab2f707681a215b5 : SUCCESS in 21m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fd4d58cff6cd4ca8a1346ae337009c59 : SUCCESS in 21m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e2d86393e8bc471388d898a70c5848d3 : SUCCESS in 21m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ff80f22fe0c4e4ca1f3e46b051a780f : SUCCESS in 8m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c08eaab657594b128e29007f022dfaf4 : SUCCESS in 53m 44s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/35052e6854e8498da8f997298b231c49 : SUCCESS in 58m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/861fd3023bc2499abcdc3b55761fb941 : SUCCESS in 48m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be911d4d702c4a6fbf830d148734f6a3 : SUCCESS in 40m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76f563a0c9064572a1c200fa6af3bbdb : SUCCESS in 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7acb2a19de96468f81b0bb166c1f0de5 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae92a10e05c44018855b4ede08c56644 : SUCCESS in 26m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43ddd402af6c401aae00d7eae0b6a6f4 : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":1},{"id":"11b2b0ad9dc7ffdea43ba2813d49a227a1244b0d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:29:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5e0aac4fcc285504d8e7f42587c0e67fc853003f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 17:42:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a7c01416354322a35a71c78eb7868c : SUCCESS in 1m 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/541b8a71811e4b508c5454240e7f1d87 : SUCCESS in 1m 09s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4e614a7a9b0f4e3aaa4ef7d86d1acd45 : SUCCESS in 1m 11s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5979b6e15f8008e14ba06c6986913ab08d4fc4c2","revisions":{"5979b6e15f8008e14ba06c6986913ab08d4fc4c2":{"kind":"REWORK","_number":1,"created":"2021-11-30 01:19:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/819742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/819742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/819742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/819742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/819742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/819742/1"}}},"commit":{"parents":[{"commit":"8bda234e40a28ab5b526ae4f62b0e26c5ba66e29","subject":"Merge \"Make arm jobs voting (but not the pipeline)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bda234e40a28ab5b526ae4f62b0e26c5ba66e29"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-11-30 01:19:17.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-11-30 01:19:17.000000000","tz":-480},"subject":"account: Remove logging translations","message":"account: Remove logging translations\n\nChange-Id: I1c415ec6fcfe426f71267c2fb7d714c2d84e0014\nPartial-Bug: #1674543\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5979b6e15f8008e14ba06c6986913ab08d4fc4c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5979b6e15f8008e14ba06c6986913ab08d4fc4c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
