)]}'
{"id":"openstack%2Fswift~820855","triplet_id":"openstack%2Fswift~master~Id8cd84e6dc6e8775adb346c1841d5b74f91d7e04","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-12-10 09:06:04.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id8cd84e6dc6e8775adb346c1841d5b74f91d7e04","subject":"Unit test logging of object updater timeouts","status":"ABANDONED","created":"2021-12-07 11:55:24.000000000","updated":"2021-12-10 09:06:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c5707b4601fb8218256f8ab6400dad9c458ea6a4","_number":820855,"virtual_id_number":820855,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-07 15:05:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-07 15:05:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05585423a55c0cdbb5467e9338509e56d4e9d6af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-07 11:55:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc7932cd2bf6ba8803539e84894858c0c060822a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-07 11:57:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"6db41065983a5f17e753fbf71154223117df1d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 15:05:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c8a43214574c2384f478c136a1fc83 : SUCCESS in 9m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ee3dcc191c942b894a38600bb774597 : SUCCESS in 42m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b7bf86a6f204ab78be996b9c5a2f270 : SUCCESS in 38m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b86e031e33d14f6a86323ae17e1cc8f6 : SUCCESS in 38m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/34332a2daf2e4ffcb0cb3dbb93594ae2 : FAILURE in 16m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/05ce2581796d4541826f43d5bee4ce57 : FAILURE in 15m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b321a8d1e4e4d4b995e7109933a934a : FAILURE in 16m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/61e26a901dd24471a50d3b7288036319 : FAILURE in 15m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8d91ad73d3b143f68718c5648e2af5c4 : FAILURE in 19m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/007fb2dbc71c40af926c55a57bf97818 : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/727e6afe02434c98b37e31caec21e745 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0570a1f89e0646119965006ef30090b2 : SUCCESS in 18m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/554fc6e2e74d498cbf93fd900036eac0 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a54615c621394633b414e7f1319db9e6 : SUCCESS in 20m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ec674b1ef0b1491c875876dcc452d703 : SUCCESS in 22m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b7578f5320a405393d161ea0b130203 : SUCCESS in 30m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e77d6ed9670d45f2bb3048b1cffa317b : SUCCESS in 40m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b69d3bde80460a9522330f13c56ff5 : FAILURE in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef6941b359c94096a114d8dd0c91f4f7 : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68e9b834a57041f49f44bc93308c6627 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a3da4948cf0b69c57dfe3c71642f5dea1c47d4f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 23:04:36.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d02daa6ec69e4b10ac80a705dd167343 : FAILURE in 54m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f717bd24bf014f9091783bc276e0489c : FAILURE in 55m 11s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/22b7d736b91540cb8e7758af3c7b271c : TIMED_OUT in 3h 01m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/22c4b13eead846a896f9323f09819f83 : SUCCESS in 1h 09m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/961a296405bd4860b6b96dd678ce9200 : TIMED_OUT in 1h 31m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c5707b4601fb8218256f8ab6400dad9c458ea6a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-10 09:06:04.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b19875a03d4970eb85f11e4319b8565afedf29c","revisions":{"0c984704f652d8ab106a7905e5d0de6e06e3ccc3":{"kind":"REWORK","_number":1,"created":"2021-12-07 11:55:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/820855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/820855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/820855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/820855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/820855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/820855/1"}}},"commit":{"parents":[{"commit":"bc8377f989e113af8c44454117aa3345b68b395f","subject":"updater: Add timing stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc8377f989e113af8c44454117aa3345b68b395f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-07 11:54:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-07 11:54:16.000000000","tz":0},"subject":"Unit test logging of object updater timeouts","message":"Unit test logging of object updater timeouts\n\nChange-Id: Id8cd84e6dc6e8775adb346c1841d5b74f91d7e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c984704f652d8ab106a7905e5d0de6e06e3ccc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c984704f652d8ab106a7905e5d0de6e06e3ccc3"}]},"branch":"refs/heads/master"},"7b19875a03d4970eb85f11e4319b8565afedf29c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-07 11:57:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/55/820855/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/820855/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/820855/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/820855/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/820855/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/820855/2"}}},"commit":{"parents":[{"commit":"094fb585faf31849c40da98c0e4832161f3e42c8","subject":"updater: Add timing stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/094fb585faf31849c40da98c0e4832161f3e42c8"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-07 11:54:16.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-07 11:57:42.000000000","tz":0},"subject":"Unit test logging of object updater timeouts","message":"Unit test logging of object updater timeouts\n\nChange-Id: Id8cd84e6dc6e8775adb346c1841d5b74f91d7e04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b19875a03d4970eb85f11e4319b8565afedf29c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b19875a03d4970eb85f11e4319b8565afedf29c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
