)]}'
{"id":"openstack%2Fswift~821020","triplet_id":"openstack%2Fswift~master~If47e6c824b74b4f37b2edb4208bddb79bd5fa246","project":"openstack/swift","branch":"master","topic":"bucket-ratelimiter","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-12-08 22:24:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If47e6c824b74b4f37b2edb4208bddb79bd5fa246","subject":"updater: un-pickle in _iter_async_pendings","status":"ABANDONED","created":"2021-12-08 12:52:49.000000000","updated":"2021-12-08 22:24:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9be8ccd085c9cf9ef7b31392b8c3d0ef3443d21","_number":821020,"virtual_id_number":821020,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-08 14:14:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 14:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e9bb7a0635e62d226feb3e78799e7a3e887bfc2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-08 12:52:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8c578fb5774776ae8a9aa7da2173ec3e97a579a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 14:14:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e392d991603b4a15a6ead0e2f3c8c37a : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/a595359e10ad4342b998e809e0b56b05 : SUCCESS in 57m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52dd1de6905645f687369c8ec4c36e8f : SUCCESS in 45m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f97f174102c3480ea250cb6026eaea9a : SUCCESS in 46m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c24c28618c7e4cd69c65deaa91ab1995 : SUCCESS in 9m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/915e4b539b144f84b15eb189750e0315 : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9cd470dfbd649f5bc553df98956b1ba : SUCCESS in 23m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4048f77a148946829c26cca0cb1c8e07 : SUCCESS in 15m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1e7baa641b1e4eaca225ae68a11150b8 : SUCCESS in 12m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aaf5a5fe9fe74751ad82cf907013aa2e : SUCCESS in 16m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/48bc202e80e74eebb73dd797d46d31b0 : SUCCESS in 25m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a07958c494574afa99db5fa4302a008d : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/233579d274a24d0c8102a3aa69d9022e : SUCCESS in 27m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c46b63e306a54065a4dccfd340941db9 : SUCCESS in 20m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d19e21efe3b4c59859a08e1760779ef : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/672df3824f10442fa411b5593f2e5f14 : SUCCESS in 26m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fd589bb0922f44b9b95475661a0ac329 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d647d087e9ca4107b99946ea373b3097 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b7becccad154ceda3978fe8b906cbd3 : SUCCESS in 25m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa0e6c48531a4688910a6c9ab348a5eb : SUCCESS in 1h 14m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d4cfd97649674c61b602b3ea088ec9f3 : SUCCESS in 59m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/effc77b002744f94842a86a34cca194b : SUCCESS in 43m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cccb407c7f08446188388c1397f6c7d9 : SUCCESS in 34m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66f7ad83328240189f975959799f0877 : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08e4766df08451eb1951b095e4eeba0 : FAILURE in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/feafc35bec704ff2abf2fbc94c421c51 : SUCCESS in 35m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d2c8ad5c0174076ac13ab7532890d3f : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"49ff98d84551d4864d6ba7789a45d981f6c92d67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 15:09:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2bc3e4dc9944f3daa538e77fd32495d : SUCCESS in 43m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2debf2674b2748809bd81967596b44a3 : TIMED_OUT in 1h 31m 45s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/0698e0eced9248a7bff0bc4297b0fcb4 : FAILURE in 2h 13m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e53cd782f37a4f618d6aa41780983237 : SUCCESS in 46m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/09b56927d6024789b8007da5dacbc129 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b9be8ccd085c9cf9ef7b31392b8c3d0ef3443d21","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-08 22:24:10.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/820962","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e96a19e3e4a9cbc07d7d297e1eeba27c1b7a3e89","revisions":{"e96a19e3e4a9cbc07d7d297e1eeba27c1b7a3e89":{"kind":"REWORK","_number":1,"created":"2021-12-08 12:52:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/20/821020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/821020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/821020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/821020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/821020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/821020/1"}}},"commit":{"parents":[{"commit":"36fadb2ac356816736288567e678cc6837fae54f","subject":"Finer grained ratelimit for updater","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36fadb2ac356816736288567e678cc6837fae54f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-08 12:22:10.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-08 12:33:12.000000000","tz":0},"subject":"updater: un-pickle in _iter_async_pendings","message":"updater: un-pickle in _iter_async_pendings\n\nChange-Id: If47e6c824b74b4f37b2edb4208bddb79bd5fa246\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e96a19e3e4a9cbc07d7d297e1eeba27c1b7a3e89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e96a19e3e4a9cbc07d7d297e1eeba27c1b7a3e89"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
