)]}'
{"id":"openstack%2Fswift~821088","triplet_id":"openstack%2Fswift~master~I11d9e08b7c44a7ca5006ed2fa00258068db546d8","project":"openstack/swift","branch":"master","topic":"break_down_global_limited_nodes","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-09-22 17:39:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I11d9e08b7c44a7ca5006ed2fa00258068db546d8","subject":"simplify signature of load_memcache","status":"ABANDONED","created":"2021-12-08 15:21:40.000000000","updated":"2022-09-22 17:39:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"02b116a5a71469066a0a9ecf19d6ba8668609f6c","_number":821088,"virtual_id_number":821088,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-08 16:59:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2021-12-08 19:30:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-08 16:59:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-08 19:30:32.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"7e2f8b7ef449dffc9937fd0e87e3fdb684d6de5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-08 15:21:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38093950029a0e90d2588ea9384a930ab01bc125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:59:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709df069a309431ca70881acad3d7b63 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/d811639ed4e648c098ea0b2cf4b7ea95 : SUCCESS in 1h 11m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e81998b061514fc6a9cc5537e2074eb0 : SUCCESS in 44m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/569ceec1ff7041c4a8755c1bbf19ebcb : SUCCESS in 49m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42588123ec534279a7a1f4e2dfd8c190 : SUCCESS in 13m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0909ad1d36b1447682c00599cd9c49dc : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4c615d32e0f3440db55aca24b1a9d36f : SUCCESS in 37m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b918d483eb7470193794c8e9f278e72 : FAILURE in 19m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f755af82da754c8dbb8425aeb150fa71 : FAILURE in 25m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/887fd1f2a377428794e1a3cac0541d05 : FAILURE in 24m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/efaeba1ee1f74481b7aa5176b1c4e700 : TIMED_OUT in 1h 31m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5aa65bd260ea45dfb5abcb00abe6d87f : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/caf3274f583d4daf8d6a6c0ca084f72e : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d945c32863348e292afda86f08d42d3 : SUCCESS in 28m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ffa27565a37d435aa11863d38083fe20 : SUCCESS in 21m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c3a321f2052402bad4961dc474d38dc : SUCCESS in 30m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/714f9c208bb447038f75fd805d209404 : SUCCESS in 37m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b62aedb49224458eaf90db001caa714d : SUCCESS in 9m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6850270c16d546a6a066d60eb27105e7 : SUCCESS in 25m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/958c756b98904f4492d21dbab54ab191 : SUCCESS in 51m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/31dea1ee53cc4ef38fa25a36ac22a233 : SUCCESS in 56m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3396a3605e3a4326801103a574ff25df : SUCCESS in 40m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/944761dd46164648bd5a089a5d00260a : SUCCESS in 1h 04m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/118aa9f369904a2bb28c66cae5b5259a : SUCCESS in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fea7a2fe50546d8bb72c38a8358b11e : SUCCESS in 9m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f022ae5152af4ae9a9c2208445a79a60 : SUCCESS in 22m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8259f14a03b146579a885a419cae45e3 : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":1},{"id":"423ef95f2a60b7a9b958d004ec6f689624f57a8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 18:31:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f75c9edea014cceb74ea8eab68cb01c : FAILURE in 50m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b671c535736241b996f1d1827dfd9c74 : FAILURE in 48m 15s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/41243e80be3a4537b270e87f8720e4c3 : FAILURE in 1h 57m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bfa3e1740ed431891a703833888e8cf : SUCCESS in 56m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4667f1a8d23e4b34b321697d806480cd : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c65ab58cdcc5c5a807867b3a3dbf41fac9d338b6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-12-08 19:30:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02b116a5a71469066a0a9ecf19d6ba8668609f6c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-09-22 17:39:24.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/820648","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bb20452249e8de509e4de1fb5baedbb3f9dc0b20","revisions":{"bb20452249e8de509e4de1fb5baedbb3f9dc0b20":{"kind":"REWORK","_number":1,"created":"2021-12-08 15:21:40.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/88/821088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/821088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/821088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/821088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/821088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/821088/1"}}},"commit":{"parents":[{"commit":"6fd8d6f11da615bc8d7f51a58548d32c89c6f1e5","subject":"Refactor memcache config and MemcacheRing loading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fd8d6f11da615bc8d7f51a58548d32c89c6f1e5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-08 15:12:56.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-08 15:21:19.000000000","tz":-360},"subject":"simplify signature of load_memcache","message":"simplify signature of load_memcache\n\nDrive-by: cleanup unused stuff in middleware test module\n\nChange-Id: I11d9e08b7c44a7ca5006ed2fa00258068db546d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb20452249e8de509e4de1fb5baedbb3f9dc0b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb20452249e8de509e4de1fb5baedbb3f9dc0b20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
