)]}'
{"id":"openstack%2Fswift~821192","triplet_id":"openstack%2Fswift~master~I9c5a46a68d544aa4bdc15bb7b4d61046eb8c0e7e","project":"openstack/swift","branch":"master","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-02-08 09:33:06.000000000","reason":"Thomas Goirand replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-01-26 16:50:53.000000000","reason":"Thomas Goirand replied on the change"}},"removed_from_attention_set":{"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2022-02-08 09:33:06.000000000","reason":"Thomas Goirand replied on the change"}},"hashtags":[],"change_id":"I9c5a46a68d544aa4bdc15bb7b4d61046eb8c0e7e","subject":"Add uwsgi binaries and configuration files","status":"NEW","created":"2021-12-09 10:07:36.000000000","updated":"2022-02-08 09:33:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":23,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"e0fe5b2eaaefe48b3a7a84ef9d76b315891e00db","_number":821192,"virtual_id_number":821192,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-26 17:54:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 11:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 18:26:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-02-01 22:58:11.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"2544f1d5b3774c69b58fffc4d05b6695247ec12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-12-09 10:07:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3485d6ffe0e7c0c43871f7c1064221eac77bb107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 11:12:10.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b71c3c7333341d7a7f7e648561e2f3b : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/fa2810fc61e74590b83ddd2c8f1836ba : SUCCESS in 1h 02m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/942f3755959b4fc790bf7775f1c55052 : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/415ec7185eca428c81afaaab9eb7b550 : SUCCESS in 46m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/299cd878a67941919863da0bce1c4a80 : SUCCESS in 10m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aae6f91e86c64a5ba6361007462348b7 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bba9ea1a243747ec93f3e2f0bd001dfb : SUCCESS in 15m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d0b52e0184c434485d5174ccf875a20 : SUCCESS in 16m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a0bfd154144644c8b938d35e634f7540 : SUCCESS in 16m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/327016993784453ca96a8b8300bdbac6 : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ceba4934873e46e1885a4c1f9a10abd4 : SUCCESS in 16m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4612393fcedb48e5b94e657056ce4b48 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea1a496e9a0a4e589e084725af8fe36b : SUCCESS in 20m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84be4575644148d9856bef9ea5b7fc0c : SUCCESS in 17m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d47c29fec14445ad8156918fca0befb0 : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b35c93a48db44b58f7295b3f12ff32a : SUCCESS in 17m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/875dac86c9384667ab3d12bb0e10693f : SUCCESS in 20m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/79f1008afb35412abbb7375d77c2e453 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab73697abc714f79b71fa3107f80b21a : SUCCESS in 24m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/28a0388cf1a14cce95e6161b9632b991 : SUCCESS in 51m 58s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c18ac476aaef49dc946096af249136da : SUCCESS in 55m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ce9ea133fab463694a32d31a182b2e7 : SUCCESS in 40m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7115826d133c4355aba04c40a527a747 : SUCCESS in 40m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0705061785114b04b6ea7552f09eb046 : SUCCESS in 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3325e84acdb74dfe97246b0a87392605 : FAILURE in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a788cf0a12d44741ad3ca113297bdf06 : SUCCESS in 27m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9ba5ecdf43f4c16a9ad09a6743edb3f : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0e83b745591982c1476610bb010520115d41bc88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-12-09 12:24:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f395d8f36004116c8a5cf212186f9c85f0426336","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 13:44:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e716dbd8244598b0f9857b0191b087 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d60f1a782f014226ab88770f1aeffb7a : SUCCESS in 1h 12m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c7b2cd92fb54e2d8113247586d9ae95 : SUCCESS in 46m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e3b93634bac4c30a7c331b49049a2bf : SUCCESS in 56m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/63ce9e4b5c4e433c84fd07553c1672fb : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ec0e55c40be4ac6a4aeadba6b15c35a : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88569d351e4647c6855484ed3df4d000 : SUCCESS in 16m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e704912bcdce4ea3838ef5a793ca2656 : SUCCESS in 18m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bbc82122a8c4c618c984653a8973d27 : SUCCESS in 16m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/57ba16f865244ff29247e4db1d2257f5 : SUCCESS in 18m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10c59cfcdae444b1834ea996fd35a8a2 : SUCCESS in 14m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/92d68c8cab5048daa33cb9f1cba673e2 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0dc9ba0652da487eade61a4ca2cd4ab2 : SUCCESS in 19m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e582de076b5b43f98adfba6fe73a78f1 : SUCCESS in 21m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/40d23f3b095e4f4fa7f30b2e0c0776d4 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4acf83f7021e4336a2495a7f241d3c83 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3332edfabdd04b508d43a4fff48f5748 : SUCCESS in 30m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d507228db8aa4ea6a6ba9ff337ad9135 : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4f81ec935ba945598d30bb4988a88b86 : SUCCESS in 23m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5539c6d7b9b5426586134f66d2bc163d : SUCCESS in 57m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/372c4a4a2d8b4fd185d416436c2f7915 : SUCCESS in 57m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5b3895248864dd4aabf43c526603d0d : SUCCESS in 48m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f1d0aa332f9462e81322dac690b4dc1 : SUCCESS in 42m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcbf30733b0c429180ad65b3b5996f3c : SUCCESS in 15m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5346cfcc7b2d4f99a4b119d7e4b7d4d4 : SUCCESS in 7m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b598cd180c346459e10236f288fc3bc : SUCCESS in 22m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7832a47098624c0fb54fee72adb6c056 : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":2},{"id":"0a4d94d4fc851610417ca15beae0748f0c92ea90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 14:49:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4895f28a2e404e399121bb2dda527ea1 : SUCCESS in 47m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a12485530d149238532a699e598f3fe : SUCCESS in 45m 39s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/a3fc63bacb4140ffad2af6e93a5f5864 : FAILURE in 2h 19m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eae432621ae4344bc8b0ba969e552dd : SUCCESS in 1h 01m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c8fd5aa20764aa486cf3fdbcba3ba02 : SUCCESS in 52m 53s","accounts_in_message":[],"_revision_number":2},{"id":"6cf8a27af33a0045d6d752efc46d1e2ea9f2fbb9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-10 18:26:54.000000000","message":"Patch Set 2:\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b376761315b8af169f16203c36434d01abedd902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-26 16:39:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab4300f666cf72ec40f58849e6b3b7d3c889c1f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-26 16:50:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a22776d0edb4cefe571783cd79cd2d3aba21408e","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-01-26 16:50:53.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8ec4fbb301aa2c42878cdada9167f3c386af6ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 17:54:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f74db00dd44304b3cafafa0f0735ea : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/ac11865c63984752974ecc860e12b144 : SUCCESS in 58m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f129611fab44efb84acf0977f060f11 : SUCCESS in 47m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd24ded4070444cb8f6d87971aa54c0e : SUCCESS in 46m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/264522dce604485986f545780bb20b68 : SUCCESS in 8m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b444a6a1e41e4d169690d02c7b03f79f : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6847a8ffda1040e2a0024d97500e9b86 : SUCCESS in 16m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eabb9e8fed534ac7821c592176aee693 : SUCCESS in 16m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35f0b4db82d64fceae54c7db73c5636b : SUCCESS in 19m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/47d84a1a02d1438f8bc50276451fffff : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/406bf6b5ace549d69a4083f14bff2cb9 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/967c71a609e343539e639fd39a40b2d8 : SUCCESS in 17m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/094a2dc6f26946a1a68d56b3e3635e60 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a62c56ff3ef44e31a93998e0b87aff65 : SUCCESS in 15m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/57f157079d2249e88e61f70b05254ef7 : SUCCESS in 20m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d193fc18af5d4d08a1ebf0051f60fe05 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc1a66492a694241927022f86049314e : SUCCESS in 19m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/97cda974793943a1b1f372cbb7d82213 : SUCCESS in 48m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e5bd860c33241509300220b1c97d462 : SUCCESS in 1h 01m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74113b7ff33b408d8f56d7f1f208ea5f : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0f1d0c33e3f41aab5934156358b6bea : SUCCESS in 42m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5661166468564119802acba54c744eb9 : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15496ab7290642bcaaa88bc97b6b019a : SUCCESS in 8m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad8fd07d31eb4ad48d481dd5699dd665 : SUCCESS in 20m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ad9700a5c604dbcbbea018108462fe5 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":4},{"id":"90f1e4d78e0c19c7beb6ae9c7c66407ed19f9e5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 20:00:25.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c5cee91dee14c10adfd8c6630eec4d7 : SUCCESS in 31m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74f6cbab0ddf4a9ab7399cae25aaf979 : SUCCESS in 35m 02s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b3ec2ab07bf845f4975e780b683872af : TIMED_OUT in 3h 01m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/528fc99855cf4eda8b15d8dcf6dcfe92 : SUCCESS in 36m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec30afd2c2f0452ebfc185766262bd37 : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":4},{"id":"e8903859d90e3854acff9bf7900fb38c25601852","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-02-01 22:58:11.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e0fe5b2eaaefe48b3a7a84ef9d76b315891e00db","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2022-02-08 09:33:06.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"124099d4ee8f8c6126325f49dab5759efbc842f4","revisions":{"832ff35ea138d85448794d1c91f5e44a446f7395":{"kind":"REWORK","_number":1,"created":"2021-12-09 10:07:36.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/92/821192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/821192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/821192/1"}}},"commit":{"parents":[{"commit":"1859f2e161f396023d8328489fb4b88c303797a9","subject":"Merge \"reconstructor: include partially reverted handoffs in handoffs_remaining\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1859f2e161f396023d8328489fb4b88c303797a9"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-09 10:04:02.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-09 10:04:02.000000000","tz":60},"subject":"Add uwsgi binaries and configuration files","message":"Add uwsgi binaries and configuration files\n\nAfter many months of using Swift over uwsgi, it is porbably a good idea\nto simply share these, and make it a Swift standard. Indeed, some of the\nconfiguration isn\u0027t exactly easy to guess, and require a lot of trials\nand errors to get it right. For example, the directive:\n\nroute \u003d .* addheader:Date: ${httptime[]}\n\nis necessary for tempurl to work and took us a lot of time to find. The\n\"listen\u003d100\" and \"threads\u003d12\" took also some time to benchmark and get\nright. How about making it standard, and hopefully, get Swift to run its\nCI with it in a near future?\n\nAlso note that there\u0027s currently no way to get the swift-object daemon\nto run under uwsgi: it\u0027s currently broken and would need attention.\n\nChange-Id: I9c5a46a68d544aa4bdc15bb7b4d61046eb8c0e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/832ff35ea138d85448794d1c91f5e44a446f7395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/832ff35ea138d85448794d1c91f5e44a446f7395"}]},"branch":"refs/heads/master"},"d546159f9d5f0cb4733df9c068e2611879eb519e":{"kind":"REWORK","_number":2,"created":"2021-12-09 12:24:52.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/92/821192/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/821192/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/821192/2"}}},"commit":{"parents":[{"commit":"1859f2e161f396023d8328489fb4b88c303797a9","subject":"Merge \"reconstructor: include partially reverted handoffs in handoffs_remaining\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1859f2e161f396023d8328489fb4b88c303797a9"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-09 10:04:02.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-09 12:24:41.000000000","tz":60},"subject":"Add uwsgi binaries and configuration files","message":"Add uwsgi binaries and configuration files\n\nAfter many months of using Swift over uwsgi, it is porbably a good idea\nto simply share these, and make it a Swift standard. Indeed, some of the\nconfiguration isn\u0027t exactly easy to guess, and require a lot of trials\nand errors to get it right. For example, the directive:\n\nroute \u003d .* addheader:Date: ${httptime[]}\n\nis necessary for tempurl to work and took us a lot of time to find. The\n\"listen\u003d100\" and \"threads\u003d12\" took also some time to benchmark and get\nright. How about making it standard, and hopefully, get Swift to run its\nCI with it in a near future?\n\nAlso note that there\u0027s currently no way to get the swift-object daemon\nto run under uwsgi: it\u0027s currently broken and would need attention.\n\nChange-Id: I9c5a46a68d544aa4bdc15bb7b4d61046eb8c0e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d546159f9d5f0cb4733df9c068e2611879eb519e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d546159f9d5f0cb4733df9c068e2611879eb519e"}]},"branch":"refs/heads/master"},"90e9f1a9fa987e4acbb3eaa9da101088b89fa71d":{"kind":"REWORK","_number":3,"created":"2022-01-26 16:39:54.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/92/821192/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/821192/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/821192/3"}}},"commit":{"parents":[{"commit":"1859f2e161f396023d8328489fb4b88c303797a9","subject":"Merge \"reconstructor: include partially reverted handoffs in handoffs_remaining\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1859f2e161f396023d8328489fb4b88c303797a9"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-09 10:04:02.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-26 16:36:33.000000000","tz":60},"subject":"Add uwsgi binaries and configuration files","message":"Add uwsgi binaries and configuration files\n\nAfter many months of using Swift over uwsgi, it is porbably a good idea\nto simply share these, and make it a Swift standard. Indeed, some of the\nconfiguration isn\u0027t exactly easy to guess, and require a lot of trials\nand errors to get it right. For example, the directive:\n\nroute \u003d .* addheader:Date: ${httptime[]}\n\nis necessary for tempurl to work and took us a lot of time to find. The\n\"listen\u003d100\" and \"threads\u003d1\" took also some time to benchmark and get\nright.\n\nAlso, to get chunked transfer work properly, this is mandatory in the\nproxies:\n\nplugins \u003d python3,transformation_chunked\nroute-run \u003d chunked:\n\nThis was also hard to find, as badly documented.\n\nTo try this, simply run something like this:\n/usr/bin/uwsgi_python39 --http-socket 10.0.0.1 \\\n\t--processes 20 \\\n\t--ini /etc/swift/swift-proxy-server-uwsgi.ini\n\nNote that the Debian package simply takes the number of processes from\nthe proxy-server.conf (this is done in the startup script).\n\nHow about making all of this standard, and hopefully, get Swift to run\nits CI with it in a near future?\n\nAlso note that there\u0027s currently no way to get the swift-object daemon\nto run under uwsgi: it\u0027s currently broken and would need some love.\n\nChange-Id: I9c5a46a68d544aa4bdc15bb7b4d61046eb8c0e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90e9f1a9fa987e4acbb3eaa9da101088b89fa71d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90e9f1a9fa987e4acbb3eaa9da101088b89fa71d"}]},"branch":"refs/heads/master"},"124099d4ee8f8c6126325f49dab5759efbc842f4":{"kind":"REWORK","_number":4,"created":"2022-01-26 16:50:27.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/92/821192/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/821192/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/821192/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/821192/4"}}},"commit":{"parents":[{"commit":"1859f2e161f396023d8328489fb4b88c303797a9","subject":"Merge \"reconstructor: include partially reverted handoffs in handoffs_remaining\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1859f2e161f396023d8328489fb4b88c303797a9"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-12-09 10:04:02.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2022-01-26 16:50:16.000000000","tz":60},"subject":"Add uwsgi binaries and configuration files","message":"Add uwsgi binaries and configuration files\n\nAfter many months of using Swift over uwsgi, it is porbably a good idea\nto simply share these, and make it a Swift standard. Indeed, some of the\nconfiguration isn\u0027t exactly easy to guess, and require a lot of trials\nand errors to get it right. For example, the directive:\n\nroute \u003d .* addheader:Date: ${httptime[]}\n\nis necessary for tempurl to work and took us a lot of time to find. The\n\"listen\u003d100\" and \"threads\u003d1\" took also some time to benchmark and get\nright.\n\nAlso, to get chunked transfer work properly, this is mandatory in the\nproxies:\n\nplugins \u003d python3,transformation_chunked\nroute-run \u003d chunked:\n\nThis was also hard to find, as badly documented.\n\nTo try this, simply run something like this:\n/usr/bin/uwsgi_python39 --http-socket 10.0.0.1 \\\n\t--processes 20 \\\n\t--ini /etc/swift/swift-proxy-server-uwsgi.ini\n\nNote that the Debian package simply takes the number of processes from\nthe proxy-server.conf (this is done in the startup script).\n\nHow about making all of this standard, and hopefully, get Swift to run\nits CI with it in a near future?\n\nAlso note that there\u0027s currently no way to get the swift-object daemon\nto run under uwsgi: it\u0027s currently broken and would need some love.\n\nChange-Id: I9c5a46a68d544aa4bdc15bb7b4d61046eb8c0e7e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/124099d4ee8f8c6126325f49dab5759efbc842f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/124099d4ee8f8c6126325f49dab5759efbc842f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
