)]}'
{"id":"openstack%2Fswift~821216","triplet_id":"openstack%2Fswift~master~I661cf7cb58a70a66de52578caf051f05e14bab97","project":"openstack/swift","branch":"master","topic":"bucket-ratelimiter","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-12-09 15:15:56.000000000","reason":"Alistair Coles replied on the change"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-01-05 10:27:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I661cf7cb58a70a66de52578caf051f05e14bab97","subject":"updater ratelimiting: fix tests to use the expected path","status":"ABANDONED","created":"2021-12-09 14:02:37.000000000","updated":"2022-01-05 10:27:59.000000000","total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"0c3d49a2bbb289ccffd7a1d68f4981b0bb3f2640","_number":821216,"virtual_id_number":821216,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-09 15:08:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-12-09 14:40:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 14:40:38.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-12-09 15:08:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f1896cf39794773b2c6daabe9e25fd451986d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-09 14:02:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6079d3f05cf89d807f67f69eec89ac5af934b94f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-09 14:25:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48e93a824cfc8decdf7f59303b74f075f809c8b7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-09 14:40:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"99bcf07df6e31c3fde861ac1140e46702dd3b149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 15:08:06.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b774607e5da64bcca44f5c50a9a530cc : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/22d2dc29f2cd4bad974446e3343c96b6 : SUCCESS in 55m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51575762d9254cf7ae89b5c45320636b : SUCCESS in 42m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cfd6c91e69141cc9f59134aea298681 : SUCCESS in 39m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c1090d0b08941f1b8f4ba2c530d85d1 : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/412e21205c0744e78d9ece4886b36974 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/072998f094ed40daa597b40c2a1b553b : SUCCESS in 19m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/524e9ef3cda743e0a91edaf509a936dc : SUCCESS in 15m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0b6afe0f00684a0ba9e9ecea6d77615a : SUCCESS in 14m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/22674d3faa48424ea773cb317e105c77 : SUCCESS in 14m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4fd931a31e44740a79042b7a69178f3 : SUCCESS in 14m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/324031ead2c045a3a9dfbd684a24126c : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e9396a677fa4a268e8f4cbbf892a853 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/db439360dd174915a77074b4e53e7500 : SUCCESS in 16m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/defaa698ee4641bcbf4e8d8258a0c5e1 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/417a0740a4f14c2c84efcd1ca37e1525 : SUCCESS in 24m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/78dc86a200874e46acd4ef80257bdf34 : SUCCESS in 20m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bd66ee29dd5e40158e99d2b3134a448d : SUCCESS in 7m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1215c6e12d1421d96433673b50dfde6 : SUCCESS in 19m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f5935334a41848389b8252cbbcc073e8 : FAILURE in 49m 08s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5baa801c5c4b4116a35e1ad950d2bc28 : SUCCESS in 1h 02m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acdeb6f8114c4eafb460b78aa83248d3 : SUCCESS in 40m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37aa3e860c3e4ecf8da60ad0af3e63d4 : SUCCESS in 41m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e14532b8cbe04c91950d7eadc841d34a : SUCCESS in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ae200e5e94d42c7abcacf9a3e8aab77 : FAILURE in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/671cfd405d3e4bb8b3fda8b215b99431 : SUCCESS in 20m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a11b29c605f4e7ca841a5a7139b8534 : SUCCESS in 48m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f78720a1f16972abaf513d6caa064ad7b52bd214","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-09 15:15:56.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e61bb1593f372254d935fe553e2335f3bcc164e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 16:32:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bed3fec914174f179825a0f056c32970 : SUCCESS in 1h 00m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d48a31cc0ab4615a7dd59613b6f27f5 : SUCCESS in 58m 18s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/46d53c27a09042e1bd958a91130adbc9 : FAILURE in 2h 22m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/48bb78066c0b47059001146fc43010fc : SUCCESS in 55m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a22d45053a104ff5a4ecaae23cdd163d : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"0c3d49a2bbb289ccffd7a1d68f4981b0bb3f2640","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-05 10:27:59.000000000","message":"Abandoned\n\nsquashed to parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebe570d1a434f98d8a70a025ab5aaa2ca7c5f819","revisions":{"ebe570d1a434f98d8a70a025ab5aaa2ca7c5f819":{"kind":"REWORK","_number":1,"created":"2021-12-09 14:02:37.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/821216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/821216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/821216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/821216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/821216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/821216/1"}}},"commit":{"parents":[{"commit":"e02b5c3f737a53c3075f511da0a5381968716b98","subject":"Finer grained ratelimit for updater","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e02b5c3f737a53c3075f511da0a5381968716b98"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-09 14:00:20.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-09 14:00:20.000000000","tz":0},"subject":"updater ratelimiting: fix tests to use the expected path","message":"updater ratelimiting: fix tests to use the expected path\n\n- fix md5 pep8 violation\n- make all tests use realistic async pickles so that the expected path\n  through the rate limiter works\n- fix mock timings broken by the rate limiter also calling time()\n- mock out recon dumps to remove noise from test logs\n\nChange-Id: I661cf7cb58a70a66de52578caf051f05e14bab97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebe570d1a434f98d8a70a025ab5aaa2ca7c5f819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebe570d1a434f98d8a70a025ab5aaa2ca7c5f819"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
