)]}'
{"id":"openstack%2Fswift~821422","triplet_id":"openstack%2Fswift~master~I269883db03f9a735b43f3a0d801a3e07078ea019","project":"openstack/swift","branch":"master","topic":"update-defer","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-12-10 23:04:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I269883db03f9a735b43f3a0d801a3e07078ea019","subject":"fixups for updater deferrals","status":"ABANDONED","created":"2021-12-10 16:54:16.000000000","updated":"2021-12-10 23:04:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc070afadada4652004d76f01ea262075fb13091","_number":821422,"virtual_id_number":821422,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-10 18:31:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-10 21:15:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 18:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 21:15:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"96dc2a45d019ff2d729f3100e4f4a59437bade2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-10 16:54:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e91cf3f904f9a73194a99adcf840b53566b99ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 18:31:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16886c3a4674964b4e5806a712c6687 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e2afb8ea53c0475387703147a35b76df : SUCCESS in 1h 07m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9c7444661b742349cb35cecf9f99df7 : SUCCESS in 47m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc1916bc2ac840f5bf2c8d88d23069ce : SUCCESS in 47m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b9fd138d7e0e4ae9960c7346c2714ccd : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8f30bbe7cf9f42dda941bf545b0e1772 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4321fcbcd1a84cd0a9b10bc3b6871336 : SUCCESS in 16m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3da8d4c7a37a47fc8a9ab6a8605c20ca : TIMED_OUT in 1h 31m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d817c1d2bff148c28b94e435a6e92c7a : SUCCESS in 15m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f5a0ad57bc6d41308fc43a2d7f287efa : SUCCESS in 15m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2699d16b15e4401597188e02c12295d5 : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a3ceead18674cd1a80bcc070e5a9628 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db344c5111134e7f9b54a3b23dce69b6 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a8ecf8d0c3f4593b17b4dcc03b98446 : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d6d37248ff644299a2184d76a19f892 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/840c8786228140d1ae8a211f52da7287 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6b1ecf70a51744fc8549ddf6fd024128 : SUCCESS in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c77765b8c054a3e8a6f2da04d033800 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/73eafca8b8524cc898fcf781fdd4187c : SUCCESS in 19m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1ec65ccbc434214ae6e3af7312bd864 : FAILURE in 49m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a0d9088d454f4bc698d06a88dca74b12 : SUCCESS in 52m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d21848c2bb4c497c89c8d954ff93bd0f : SUCCESS in 38m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/639d201491df4e3083fa8d2a044fbba8 : SUCCESS in 40m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd5712e12a3648cfa03aa58b53f9efcb : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e66b8add0245cb81f14447d5e8bdb6 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/167344ecd79244ed98993584630b5902 : SUCCESS in 20m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4c006b829204837b857b53936d717cb : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":1},{"id":"57fdc81e7392f511d738456730ca9dbb187457c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 19:16:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e1a4296271d46219e65f2a7b9d3bc5e : SUCCESS in 50m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/191234ce5dd6412ba9124e0e7f044a55 : SUCCESS in 40m 13s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/4d6c363492794e8db0327cad32c35de5 : SUCCESS in 2h 19m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2f51ee583774636ab4f6a1032e4c4d9 : SUCCESS in 51m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e79a02e83414b549286a0e21c1c1e2d : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0fc5a6c2c9c5bbf9c60684a4d978ddccc5b60447","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-10 21:15:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc070afadada4652004d76f01ea262075fb13091","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-10 23:04:33.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/821326","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9197a8a9169e0bdaceae435aa4d0558ce16c98af","revisions":{"9197a8a9169e0bdaceae435aa4d0558ce16c98af":{"kind":"REWORK","_number":1,"created":"2021-12-10 16:54:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/22/821422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/821422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/821422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/821422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/821422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/821422/1"}}},"commit":{"parents":[{"commit":"78b637c34c0c906987baa72f9db7ee7985dadfac","subject":"Defer object updates to the end of the interval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/78b637c34c0c906987baa72f9db7ee7985dadfac"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-10 13:51:17.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-10 16:53:53.000000000","tz":0},"subject":"fixups for updater deferrals","message":"fixups for updater deferrals\n\nChange-Id: I269883db03f9a735b43f3a0d801a3e07078ea019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9197a8a9169e0bdaceae435aa4d0558ce16c98af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9197a8a9169e0bdaceae435aa4d0558ce16c98af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
