)]}'
{"id":"openstack%2Fswift~821477","triplet_id":"openstack%2Fswift~master~I3de9fb217b402df214430a6cbf920d7bd7694b32","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3de9fb217b402df214430a6cbf920d7bd7694b32","subject":"CI: Move py37 and py38 jobs to experimental pipeline","status":"MERGED","created":"2021-12-10 21:50:37.000000000","updated":"2021-12-21 19:55:57.000000000","submitted":"2021-12-21 19:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821477","meta_rev_id":"900752a3ccdbec7d8b0b7fe601141a364fd4a3da","_number":821477,"virtual_id_number":821477,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 19:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-21 18:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-21 18:36:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 22:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a427ea4a3cf30333193a0d9179789eb75f6c9104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-10 21:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a22ca880e12b47b7b5fdda9f50b7dd059c32385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 22:51:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc6edb207e44a42b7fed14727a3c392 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/e2d0ed8244954a84983fc1f997bf61f7 : SUCCESS in 51m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/012bcff3b3e34c8981cb806dc193f280 : SUCCESS in 43m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ed9cb5a2f7e4f0d8724dedab88dcc6b : SUCCESS in 41m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b8f5ad7ad3747e883ce20cfa5641146 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/af78d99faf2c4c779e7d80114d565db7 : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4744788ab1aa4c4ea1c9452966e9a183 : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58cc41aa969942b0a8f679d1e85428f5 : SUCCESS in 15m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b9c8c528a4784400998756d361fff894 : SUCCESS in 15m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8c0d12749e644bac99025aadb4662dca : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/781dee2525644b7f994a3548a3c5e878 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/10d8e6a481354990982057a707b0c950 : SUCCESS in 17m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d47b7cf7895d4bc0b0b205cf1a514ae7 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/26a45f607ca34a8685c25a0dea32f1fa : SUCCESS in 17m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cd34e65f2b2d468c89dc648949bea341 : SUCCESS in 17m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f7e002eff0fe4558bbe89380478508ae : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1afee3c9b884bbfa2256ed37bc405c4 : SUCCESS in 23m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02e85bbf853343178631009dec844441 : SUCCESS in 51m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/434cfa682cbe45bf8d2be907c1de6589 : SUCCESS in 56m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5cd0d37794014b35999b0df734ea862b : SUCCESS in 40m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff61215d229c4d79b144331bccea482e : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55f62c5142f9461f863f4850e96858f7 : SUCCESS in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ce6695929fe40d5b33c4a358bd7b47e : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e09ba765ecc94e2b92e4ef7919e1472c : SUCCESS in 27m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3089fe0d20144857bfaa63be1bd4b070 : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f853fdf5d237b7f84ae7c47e874202ec6e57d2cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 23:46:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d375abf969e49c58144598ebe9ec5f9 : SUCCESS in 38m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ade99ca7c76246c9beb1c9954e027fde : SUCCESS in 38m 13s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/3514b2ff8d2e46c3bcfd60f39dbafb0c : SUCCESS in 1h 43m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbea1f33b8284c90967038f64b796d4b : SUCCESS in 41m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a6bc8bac4ea413db4a1f38cfe507e35 : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4dcb34e6e81d0e3b6d9520b8eb32a3ddddcfdb35","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-21 18:36:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fe129216984571375fea24809e831f233540558a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 18:38:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4dc06b86c4404976453ac53825558be645453fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 19:49:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440fd0873de9490582e71b48136559f2 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/84f39b143eee4888a29958d6defd4477 : SUCCESS in 1h 06m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67846c6d0ca7413d9dee02b2d0ebe655 : SUCCESS in 39m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8234aa33ea984b74a4ed2b4b351cdd76 : SUCCESS in 47m 26s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/677a00c091b64fc9a1f0986841703153 : SUCCESS in 9m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1c7056504d4b4ce39e71827ba68559df : SUCCESS in 7m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/198b03a4b28c4ff994f80b44236b34a9 : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e279a76099db46f09876f99f590457ea : SUCCESS in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b34c1120850c405599fb707a94d0bc1f : SUCCESS in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0430843dbe844ee182b0abd20f6b3f8f : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/888cea9c325d4fa1bde6dcabbd05b6fb : SUCCESS in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86817c4be9bc4cbea492a5be14f4ac1c : SUCCESS in 22m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/70554a5e31e34bf68f2869d8ac89161e : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/52136f673b8a448c827c45832060bbb7 : SUCCESS in 20m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/425279c592384bb5af78dc78fc6ba581 : SUCCESS in 19m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3730b1993e414f28bc48eb57ded99eb5 : SUCCESS in 8m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/db55431e94f74592b1788827fe9ec14d : SUCCESS in 57m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/781df7a4de9348ad80ef08d5a1119c20 : SUCCESS in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d01b2cf7677a46f7ba818b27cf136cd6 : SUCCESS in 39m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be6477e080bb4ff2a0d43defe2408ae5 : SUCCESS in 45m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482d8119f692443e86f40a38b7a9a4f4 : SUCCESS in 18m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910661a21c3a4ef08c95d628b5f07f60 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af2214273feb434a8b267626c62579c3 : SUCCESS in 38m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d78f21e6bac470f853b3176cf4af546 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bd736e39dc656472c2d70ab66f7053a91767731c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 19:49:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"900752a3ccdbec7d8b0b7fe601141a364fd4a3da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 19:55:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6318dea774f04475a2f86c3c09ef414c : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7a6c243d20a94e4b8fba90c647bca332 : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4c1806ca0e944fa4bc7482808759a4d6 : SUCCESS in 58s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6206e6fe5ad8c8c4f836ff7edd4f7e64acd3c17","revisions":{"d6206e6fe5ad8c8c4f836ff7edd4f7e64acd3c17":{"kind":"REWORK","_number":1,"created":"2021-12-10 21:50:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/821477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/821477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/821477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/821477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/821477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/821477/1"}}},"commit":{"parents":[{"commit":"1859f2e161f396023d8328489fb4b88c303797a9","subject":"Merge \"reconstructor: include partially reverted handoffs in handoffs_remaining\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1859f2e161f396023d8328489fb4b88c303797a9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-10 21:48:45.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-10 21:48:50.000000000","tz":-480},"subject":"CI: Move py37 and py38 jobs to experimental pipeline","message":"CI: Move py37 and py38 jobs to experimental pipeline\n\nThe confirmation that tests pass will still be nice come release time,\nbut the py36 and py39 jobs probably suffice to say the intermediary\nversions work.\n\nChange-Id: I3de9fb217b402df214430a6cbf920d7bd7694b32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6206e6fe5ad8c8c4f836ff7edd4f7e64acd3c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6206e6fe5ad8c8c4f836ff7edd4f7e64acd3c17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
