)]}'
{"id":"openstack%2Fswift~821478","triplet_id":"openstack%2Fswift~master~I17638d469f704e106ce691bed51e879c736ce153","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-01-08 05:35:43.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I17638d469f704e106ce691bed51e879c736ce153","subject":"CI: Add rolling upgrade job coming from stable/xena","status":"MERGED","created":"2021-12-10 21:50:37.000000000","updated":"2022-01-08 07:53:35.000000000","submitted":"2022-01-08 07:31:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821478","meta_rev_id":"e8773877e47378f07c781e66ff292282b02a75c6","_number":821478,"virtual_id_number":821478,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-08 07:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-04 18:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-04 18:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-10 22:49:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b5c4c4e3218ecc28693ed4a091b12a089edca18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-10 21:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7da4e94949e5f9b8a6007137ff4ba85efe95e16c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-10 22:47:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35aff004c583aaf87f6da9e5e3fcf1a6c8a3f068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 22:49:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25868532a15941c4991a7080175fd33b : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/f8087d84754c439c826fbab5b2725b09 : SUCCESS in 53m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/375288e7df544a0fb35a7a9166921c48 : SUCCESS in 43m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fb32779a1954e22bac27bbcf6fb9bd4 : SUCCESS in 41m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e7f6e01853e4622ab60d6c321fa34be : SUCCESS in 7m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7d1ef76669c4e4092a3a650410b0d12 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f7baa17f7f14b16bebe907ce4ad6aab : SUCCESS in 15m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f858690a2a0b472593b696212c96fca9 : SUCCESS in 14m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/895b9ab70ebe46c1a2146e9596624e3e : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63b248ffe4324ac296bbb9ba03a52f65 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21c6abd86c3b49978da0aa0dfe40d0dc : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b688f1681554359a8d0604882243524 : SUCCESS in 18m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9b5187a768e94bb0987b19f437d5dc69 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cac6a5217856407a96b8a34f38c7917f : SUCCESS in 17m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/341e74111b6f4863ae492b5ec664c92e : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/005b85fe699546e09375466d95d66041 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58977203217a4ce6a66e7af4ecb591b3 : SUCCESS in 21m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c4e6170cb49e4908ab5e6472c172eaa1 : SUCCESS in 48m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b32a33ac3f664081ab4d6034a8db0759 : SUCCESS in 54m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90112e1575fb42fe97fc8f08def109da : SUCCESS in 41m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/308a44bcf45e4e78aa0412b5ddd5972e : SUCCESS in 40m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79c6149b5928446ea7c1fa6fac43c850 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2291f17ec434b2a8bbb5a80818ec3aa : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0c09806374114ad7b20bb0823dca5f83 : SUCCESS in 28m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4159f06113024bc697bd0cb1684cfadb : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":1},{"id":"47b1d3d1d7741c8b3db79a7f171970d643c0b9b5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 23:22:04.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/14b927412b71407194bc285a83bbaba8 : SUCCESS in 16m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ac425b49ebc14925b7af515b4410e018 : SUCCESS in 31m 30s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/e6462ecdc52347bcaa16ca27678a3faf : SUCCESS in 18m 53s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/a83457a44fe4407a8e8f9c10c3a8c0ab : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/8fa0e8f06e154328b6e664814e7460ff : SUCCESS in 24m 29s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/e0c43aa0f742422699bd59db902fa524 : SUCCESS in 20m 04s\n- swift-tox-py36-centos-8 https://zuul.opendev.org/t/openstack/build/dbc25cad127144d5856526285fc4dcb7 : SUCCESS in 17m 13s\n- swift-tox-func-py36-centos-8 https://zuul.opendev.org/t/openstack/build/cc886021550b4529a12ca84fd4128afa : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/e63432ddf34b4efb9056af2fe712f4dc : SUCCESS in 22m 15s\n- swift-tox-func-ec-py36-centos-8 https://zuul.opendev.org/t/openstack/build/e4367873ce72450387f977f879a347a4 : SUCCESS in 19m 55s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/48db7011d2cf41698ee44c09bec35aab : FAILURE in 7m 43s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/206918f8515342ea80667a34b601ba92 : FAILURE in 8m 23s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/a4deb4640ed54117976a2b25e73d3d16 : FAILURE in 7m 30s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/f1305a3915d5407e81e40f8870f18d73 : FAILURE in 8m 14s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/6b78825205d1494f9c27970ea9f4b6bd : SUCCESS in 20m 19s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/b0aec383d11a4f0cb464467dd9060049 : SUCCESS in 20m 29s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/1f20ee52fe1b4662b431f27ae48c722b : SUCCESS in 28m 43s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/9f09b10bdcbb4dd3a5534e94f81c2e93 : SUCCESS in 21m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7951c467e590e4a399a7e4b50345dad546640875","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 23:44:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46c341ef45804628becfbbe7b7fa567f : SUCCESS in 35m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7bfcb2112654c61a1d74a8d243c1654 : SUCCESS in 34m 21s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/3b20595c5bdf4219a09b05c71baff2ac : SUCCESS in 1h 38m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1900e660954040379f0cd612fd162c08 : SUCCESS in 41m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd40272803b44081aa9b86399f63e9de : SUCCESS in 39m 54s","accounts_in_message":[],"_revision_number":1},{"id":"81cf35dc950adbd2f414dc800bf927618ec70da6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 17:44:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbf04bf98a6e5ddba836552caf4b7d510a2ba461","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 18:18:54.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6759e1a91739488a87b6c17f5b4c9954 : SUCCESS in 16m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8062bc7b8a243d4974f93f9954ffba6 : SUCCESS in 15m 28s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/41b6b43ee8904bb8b378eefc080c3a49 : SUCCESS in 20m 06s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/6331326854524a4aa0055898309b1a5f : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/bc904b588abf4a6f81220010c8a295e7 : SUCCESS in 22m 19s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/af56e7e542c843748279c08d4f0c479c : SUCCESS in 26m 59s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba211e78c8fa4a59989f7822dcbc4ea6 : SUCCESS in 18m 55s\n- swift-tox-func-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/420ddd2f9aeb4b6e8682a00e33d5ba38 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8185a90932a4e67b17d017a79e7cd49 : SUCCESS in 22m 16s\n- swift-tox-func-ec-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/bf86806d155b41f39b319b5583d1099e : SUCCESS in 21m 42s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/c7b219ff41e343f3b4b9fac144c0a5ce : FAILURE in 7m 47s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/85e1ffa8c74741e7abf4adff9d769283 : FAILURE in 8m 17s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/b2343b8aad7f42b7b31527c3cbe4ad4e : FAILURE in 9m 01s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/9266e58c11414e9f915dbe20aed3b26c : FAILURE in 8m 49s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/4bca4fd94f7e40779f1caa76c2e0c007 : SUCCESS in 28m 20s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/3222fd4f5cce48459f6c4ab8ef7e10dd : SUCCESS in 21m 10s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/e970a0dfefa3476898d675d2568b9358 : SUCCESS in 19m 45s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/a26c0b372b004ed4be0733063429fda3 : SUCCESS in 23m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c01ceabcd86719cc873b2f9aba88fc4e60a43192","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-04 18:54:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9895d3275d8d5b4f8b0f91857970e51c032a158c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 18:56:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"964c225b0e1c44397cdc2805e7b22707d0007dd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 20:32:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb2f0e3a36944139880fa24046295881 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/b71dcf9667dd4d5b97e4ce3443c43858 : SUCCESS in 1h 02m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3dd8e984a4b4619959a95b8d3ad8fa2 : SUCCESS in 48m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f18806a059c407e91ea43f42b2a23cc : SUCCESS in 1h 06m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f2fad2b4661d45da86b6a472e1bad01d : SUCCESS in 9m 25s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/60e663e3e29349f2baed29a21fc25b82 : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ab32846d3a24401aad68630acfb9187 : SUCCESS in 22m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6a62fad224e74262a0697ff5b91a61d7 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/50053de730064e84a2a2190dfd23c663 : TIMED_OUT in 1h 31m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5ff216f953c44d7a3c68490c706d288 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dfd94f64777f4daeb413b95c9c6e149a : SUCCESS in 21m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a138522269824238b6f90ac952de93d4 : SUCCESS in 35m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/77fdacf16cd04101ae3dd1e658ebb929 : SUCCESS in 29m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0bae9ccbc8f346dfa7bc453c81aaa674 : SUCCESS in 26m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/43c8a0ee116e430d9fb59c8d1617ba47 : SUCCESS in 29m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/63519211cd74491dadc174d25fb1c0c9 : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52ca2b699d2b4b979977783f10dd8c16 : SUCCESS in 52m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/272c4a88ab57484f9129b1eb4bc58bed : SUCCESS in 53m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/952a65100c8e470e8c65b9056cc29f82 : SUCCESS in 43m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/588c7ad493214275bb67752bfa52b13a : SUCCESS in 39m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/629e0b839fac4386a7afaff0be76241b : SUCCESS in 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aca2dacd5a649f0951f07ea3a77a94d : SUCCESS in 6m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/304faa9bb6994c64bb6be6b297ac5fad : SUCCESS in 21m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23d8c2f979fb4d139c8354d316a5385f : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fc981e6a261620551a451ef1d808be7ec9e93253","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-05 20:59:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6ff9e2331b0a127ba931a0857921578fb19df40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 22:34:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c957dd7d6ab145a9a372176b3f983f5a : SUCCESS in 7m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/325a250d640f4c29a0c1b2a6abfe2278 : SUCCESS in 48m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9a6a9cfda3c46cabd8b4410933d7eb2 : SUCCESS in 46m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d693a627d7f24afdb7688419ed23dfa7 : SUCCESS in 41m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7bc7310f166f4669af812ba51bde5390 : SUCCESS in 11m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b03066660ad94c02ae561660451c85d2 : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8b22bac56b94076b840537c055e89bb : SUCCESS in 15m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d8cba2bba3ed45f6b1348280eb8913b0 : SUCCESS in 14m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9caf0e3837074d7489b2bf7e216d3d0e : TIMED_OUT in 1h 31m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eb952f5edd644b789295e69835592a43 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c20c6c66571549b39a2b43d65c7ed98a : SUCCESS in 25m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0544f114e1d247cda5dca795d62cc1bf : SUCCESS in 29m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fcaecfa96466432fb0ea8c35092d58fa : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8f4409c45cb84f5cbab906a5d694a0d7 : SUCCESS in 24m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6cfcffe857644078a0b1adaefa12c204 : SUCCESS in 15m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6751cdc67eb3443197c3d3ea9f14c443 : SUCCESS in 7m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/48273ceeaacc447d9b4627f0247b6b27 : SUCCESS in 21m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9cc4a3556bb949d991d7e00077293cc9 : SUCCESS in 57m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/85c3458ddc7742bfa8468c59fe5a3610 : SUCCESS in 55m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/415370a0b26144478f95b7a283a7af70 : SUCCESS in 40m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f22c1ffd187641e99163e039bbbd9639 : SUCCESS in 46m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc2cdc43598e4962a1da38a6ee80750f : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b8a25c09d64ccf8c4abcd160c53a91 : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbc2a2031c1549b891a9ed8f64d9d265 : SUCCESS in 22m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b73505a7a254723a6b7b7af8a857e78 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"92e04ae156417c1a36f91ce1488fa0c5acaed79a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 22:38:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eccdddd9159341769ba6d4830300ea39 : SUCCESS in 29m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d44502e09d74a49913c37cd962d4f4c : TIMED_OUT in 1h 31m 14s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/25edc0afa79840a0802945602468320d : SUCCESS in 1h 24m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4d89e3921f840cfb27b5994c9090231 : SUCCESS in 32m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e7526db056c452e9fed98ce806b610b : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":1},{"id":"80b42383f4065ecd0277598ddaf3324522f9a28c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-06 00:20:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d8e3314e6c39c6e45df65858f020100ccd6af61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 01:54:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0948084fbac34fe1bfd30cb582a9dfec : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/9766e5d46bd749419a584b17216274d6 : SUCCESS in 54m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54afdbbaf9954ba3b4402b385957faef : SUCCESS in 44m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81f866375ca2452e935c22c9f9b8511a : SUCCESS in 43m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3153cba615bc4ca18db11897f0c416f8 : SUCCESS in 7m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/02fec03070a645df83f776b5c0265fd6 : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/09621d397ba44f5d906c8edbb5a45e39 : SUCCESS in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ee8ce0338094c0caf6995a81a8bedf8 : SUCCESS in 19m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/69e4a8e711ac46809be277a73c9113a8 : TIMED_OUT in 1h 31m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/600d3ed6bbf148eb83c4860362bccd72 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05f64cfa5d39472984ba2a76228f550a : SUCCESS in 21m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80fc43ac4c8e4add819af50076d408f4 : SUCCESS in 22m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/437a3895f9774c8792830b468dbacef9 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f441cbb19cca452d96c901bc9c8bfe25 : SUCCESS in 29m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ff51093e7a994d1b88449c830eb72bd0 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/73b06a9b3f484a408c9cf5a9151f8c35 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d9c12744e33042bcbe441ad11ce690e1 : SUCCESS in 19m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8d38d8b23851401faec21b2fdb9c94bd : SUCCESS in 57m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0a66e0ac9b5498bbb0d6386f39589e1 : SUCCESS in 59m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b4f3579e46e45229cc4311ddf87fb45 : SUCCESS in 40m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a0b82dbbb224e38b5483306f3c7bb30 : SUCCESS in 41m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/512b69017c3e4d449c4afa25f85da48b : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9d174099ff84797827977d3c38941cc : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea3600c9975b41feb5bb501dbd835996 : SUCCESS in 26m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa60dba260e949beb57a037b28682ca2 : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2e7fa1fdc35258792d079ae625e9fb750837de9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 02:57:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ee2d90d8835458bac249bef60cc3429 : SUCCESS in 40m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6ed9e904f0f45fbbb31f751de800270 : SUCCESS in 1h 18m 20s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/d64a25f64512482f8452eb88f451037a : FAILURE in 2h 34m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed6b1714e08b430f8fbb789967aa9021 : SUCCESS in 1h 26m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3a72fb841e743a495dedbd7c5559367 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":1},{"id":"95c3a72139f5aad0150b61fe0285d6512e35846b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-06 19:13:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4809beaab5b3d2af453d16c41708ba046aa316a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 21:09:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1614a8c0606c4bac96cd21f80d85d98f : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/514c744556bd468ab764b8730c5f6e83 : SUCCESS in 51m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43af94c764bf4237801898effd09c81e : SUCCESS in 43m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/699bfb2ecef24a1fb53ff3be9f4e9e72 : SUCCESS in 49m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3624a73401014fdda3fde7557a9db2ff : SUCCESS in 8m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c3642040d5c451bbcaf95d9cf792cb1 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7eca30ea1494aecbc36621d96195279 : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8748230ee5cc429988e53cd0791b70b1 : SUCCESS in 16m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/40138bf2528e44869a25999367856d86 : TIMED_OUT in 1h 31m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/914d8e85a5d44a86a4e5af29eef735da : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/996d99b800514fce9bf832c3f4787f0a : SUCCESS in 25m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2990cb283e284be99972dddacf3c7999 : SUCCESS in 23m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/02dd13f500884a5f8527822b4f3a5936 : SUCCESS in 22m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bfc87dc89fef4addbc3b717048b23799 : SUCCESS in 26m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b41cb79c74584da2b541174d6612d1af : SUCCESS in 16m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a6c1699f46de4d229be7bd6b1a4d51f9 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66ce9f1c33c14f0a9443ccaedba9e193 : SUCCESS in 20m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dc51d75f216a4ea897449b1aabd1b96e : SUCCESS in 1h 04m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/299a460465dd46468a4e78f421bb1e9a : SUCCESS in 1h 01m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd3b19fe87974973a0d95dc88fae4e99 : SUCCESS in 45m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8b9045ed30341baa5436a926e00cf45 : SUCCESS in 41m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e49d6d8010cf48f69e80f0851fbd2f78 : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3575787120924e448808912095d89b51 : SUCCESS in 8m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b26a5c54e394c6183e12fb2cd7afcdb : SUCCESS in 21m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fd66109d299480ca9b132e22525becc : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c842eb8c7c5f6635237c369e7a85a12e9ef136da","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-06 21:11:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e584be052f07f5c48c7854e65b5bd014c852757","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 21:57:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11b664bf5269452eaae6a0ed5566c318 : SUCCESS in 38m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e888656b0a3845aab641cba18ab5c28c : TIMED_OUT in 1h 31m 59s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/81f81d80774640bd8c433432b889024f : FAILURE in 2h 17m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c039940616b447e29367ece1b5996bc5 : SUCCESS in 50m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee5a030fdebc471385e1718d465a6489 : SUCCESS in 44m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2a81bca5814b91aa14f80c8455645311d6ecaef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 22:13:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45ca442077814e97b2765cfc5efb4f72 : SUCCESS in 10m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/a25d57705d564941b273d3daf82dee67 : SUCCESS in 54m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66768d96d541485881715fd0b5f9cf7c : SUCCESS in 47m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e12820466c0472897533d4b26dff64e : SUCCESS in 42m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ab90747b351c480ba4bccb3713b68254 : SUCCESS in 8m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4174ff192d1d4dcd8747ea12b34ac37d : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/112731c682134a7593157925796a1865 : SUCCESS in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f63f2e507c894c0d82fe5f4a08982f06 : SUCCESS in 16m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b1f94b208f894488a35e96eed4d2fc62 : SUCCESS in 15m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2d240a45d6a64879b220b857c0febfbe : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1846e7ec7284fd8b20fa4d5fcb86d1f : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa9126f2af57497592ccf84cae8aa145 : SUCCESS in 21m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aa39cca9571b47ed9ca0f778b71854b2 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/75fda07c66aa4185ae226453932a66a2 : SUCCESS in 22m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b40cd500b34a42d39c534da91c60bde0 : SUCCESS in 25m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad23a77214b14620a486d05f3e247346 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13a271fc810e43c29bbf4a15512bd203 : SUCCESS in 25m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c57d4dbbac1747d7ad7506e982b2a38f : SUCCESS in 51m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/63a75caa1024453b82a1cecd18e7626c : SUCCESS in 59m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0bd7c9bf5724544ad8487df5841bb3c : SUCCESS in 41m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5a1f15499f674e448d53f64b84b7c2b4 : SUCCESS in 41m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d025cc4db8d4bf9908b95dc42e2cbec : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46428625ec243c893d624763edd7119 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/439cb8e53c454dadb7433c0624c28048 : SUCCESS in 33m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c002c4c7afa4f5ea2369af568839d6a : SUCCESS in 45m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cc0b8d04b0b6d540d164d126cc369cad5f91456a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 22:14:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2efd3ad856a1333b17e1f551e85abfefa610040f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 02:10:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63a1f9f12b5400dac4c255f926b12fc : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/d03c5e5f10c44bbf8cf6e64ac0e146ae : SUCCESS in 54m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f47083e613aa4e35b144eb97918d46bc : SUCCESS in 43m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/109583816ad0492da969df5a482bb2d0 : SUCCESS in 44m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d5239753b10541c7977bf63d1740b44d : SUCCESS in 8m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d56cdc8c0c6f4b6e92f8b7f030acab9f : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/575a14d864b445889c172d1bc6486306 : SUCCESS in 19m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3df29d0d4c6a4fb8810ddcb5f4032ec0 : SUCCESS in 15m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/60e097bf6b86488e80e851cf38497f9b : TIMED_OUT in 1h 31m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/af4ad34ef4b64691a159e79a74ebadb0 : SUCCESS in 20m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c08a8189ad55458180ad6d1d6cba4e6a : SUCCESS in 25m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1cc729b7b46a448bbd92a21931460d20 : SUCCESS in 19m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f66ff55dbf004af0a61cbd474d0e56a0 : SUCCESS in 27m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6458a6664a064a6595dda0dce89adf88 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ed9b3ddb480c4a36b00b22242d6ad49c : SUCCESS in 34m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c4385b29e4af4439976ae4c26dd04d84 : SUCCESS in 6m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2aa16e398b9e42bc943666a8aaca7a1d : SUCCESS in 50m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4be1ba6656a84cd2b889bce2135608f0 : SUCCESS in 55m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3ed20a80d1c4f5aa44b11a0b23c4bba : SUCCESS in 39m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba2ef02eb81a4e919684739f1edfac07 : SUCCESS in 40m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c57c5551d8324a58a684e493feb54f70 : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c5ffa13546447dda0fd1d8eed65f34c : SUCCESS in 8m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/339a8025227f444f8536d125f8682d5a : SUCCESS in 29m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a78a8c0bb4dd432cbd8032df4f6af447 : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":1},{"id":"87b6f837cbc738a0db24b168f3224f18f61b64af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-07 19:25:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2154963a53d1a1f51e87d2b6306c719cd995efcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 20:28:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94635238d54641fca1f0595bbff3cc9f : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/8adb8db852354c3994f98624c4ede167 : SUCCESS in 58m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb058a01bca44074927b96cb6a914bd6 : SUCCESS in 53m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e3bb17721cf4edd82f85e617c2b0758 : SUCCESS in 48m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5f9b55a6d0b4321a5fc03e92de339a5 : SUCCESS in 9m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b9a66c1e73045448fcabbc9fa887ab4 : SUCCESS in 7m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4517d7f2d9b447e789fe6c779ab4c072 : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3babcfe69bcf4aa4bc3a7c300dcdb645 : SUCCESS in 26m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f0083f4b76cf4d79b40345fd94d08943 : SUCCESS in 15m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/96397a30d52f4232930e61bd768b6328 : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c1814894ed5046f080bbc37d6d5d6a38 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/22d3e23720c840a4bc9022dd60aff374 : SUCCESS in 18m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5215b2d7df1c44e1ae944b9723f6ffa1 : SUCCESS in 27m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/331c73c8d2634eaa884754bd2fd7a6ed : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4287311f1bfb48738f39e7e302a23c02 : SUCCESS in 26m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d31c5566f474618ba900684ccc1eae3 : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3d17a711f98495e9bc575d47e52cd07 : SUCCESS in 31m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3415ed30e8143bc81d88fab4b094410 : SUCCESS in 55m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e8e008f16d54b3d9680a423dbe1b09a : SUCCESS in 52m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21e772b622b24deeab24fb2b270e876f : SUCCESS in 47m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e67ecefa45a942cf96e11ba91e4eec24 : SUCCESS in 40m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b0bf229495a4b96966d804921139a12 : SUCCESS in 24m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9152b956f28245c3b4072a4dad6c077b : SUCCESS in 7m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/544669f2fcd9493a8283b87f2323a0be : SUCCESS in 35m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f5760854b394e088201456805195fc9 : SUCCESS in 48m 32s","accounts_in_message":[],"_revision_number":1},{"id":"36161da4cc1eebd58f09acc7dca3a62de11eb2a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 20:30:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75aaa8059aea19ef27af854bba50efeea777f97d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 21:43:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b3bde1037564360a4cef02c6ef81b04 : SUCCESS in 35m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bd496f8555245de9bcae9da6cde95b7 : SUCCESS in 36m 42s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/7af39fd3f4844258b9d4dbdb965856f5 : SUCCESS in 2h 12m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f122589fdd347bd800d8e71c7e34104 : SUCCESS in 44m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/88fe7000afa041b494b8aa82c3645c02 : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":1},{"id":"82e6fe0e642866fe8c543be31035f91aa427abf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 22:05:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc773bb2e8f442cb819a01db2f7da6a0 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f160146087b747f5b4e07249eaa6c3d2 : SUCCESS in 1h 12m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2dc3807060264260860e73670ae51b43 : SUCCESS in 52m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0219373d1bf404ea75cba838fa93d1d : SUCCESS in 56m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b0d3c113a51548be8ccd2ae8b7effc11 : SUCCESS in 9m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/00aea921aaf0429c9848cea1e42a389f : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2cff4df9751a49bfbd1885a6ddb68462 : SUCCESS in 18m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16dc0cd92d8448f9aaf2d5052db9fbae : TIMED_OUT in 1h 31m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a5c4634f112e4447bd54d0803bd7abfe : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/efce4208173340049e5238df81fa8285 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c30b2604f8a44a0cb56e2e4d0889b0b7 : SUCCESS in 32m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b55aa70d96274d18810a56319218b4e4 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aa2b3e87c4cb431daf98cf398c8377e7 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/661601f1c19945deafb8337bbd37d1a5 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f3f3b7c5ff94886b19f2644404b30e4 : SUCCESS in 18m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a5024f4cd0da4f4c9f46541aa4ae3215 : SUCCESS in 7m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a11e4c7025df46d99f6f65017d65a5f9 : SUCCESS in 45m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1de88c10dca944e9a7e07f1dc0bef15f : SUCCESS in 1h 03m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc02a509e2eb4930a69d288de568923b : SUCCESS in 43m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ed2d61bb44b4b56a476b5e186503664 : SUCCESS in 31m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b271023d485d42a29a662cb398d64f07 : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ccd3047a6a4e4f861012602222446e : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1812124532574c859d383831e6c65373 : FAILURE in 16m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/597eaaacbd8c439abe3bc8851f61ac00 : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9e13b2b2267853cb8f53799e06a3f2a0814dab1f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-08 05:35:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"670e6da371ce381773ba7120fc5234b0d36d61a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 06:27:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ef07d73d384bdbb216f3c032e6e311 : SUCCESS in 11m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/17cf29b278124d81a1ace74d738848f5 : SUCCESS in 47m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43e94a59b00f4a50a2d049c383e3a60c : SUCCESS in 35m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b8680ef6010440ebf2acb65af1eee31 : SUCCESS in 40m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4cc4dfcb22954df69c36da4790ec1d05 : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b29246a517044eaa9b94cf39785e0c1 : SUCCESS in 7m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45aa3f4be9f84017acc39eecf794f598 : SUCCESS in 20m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8059ccb471e8486080c57fc249fcda29 : SUCCESS in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd06810dc9a6407cbe7e83429a34ef93 : SUCCESS in 14m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e8cb578bec14fb4841808dd5e7df827 : SUCCESS in 23m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0478afff7d041c189f1ff2d176d8e68 : SUCCESS in 22m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fb10d34b196d407ab7723fff16629e79 : SUCCESS in 28m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b4a5cc4cc64e4eeeb33148b2b4ffaa33 : SUCCESS in 46m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9be0feb4937c400b91b85e45c8a47102 : SUCCESS in 30m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4c9148444af1432cbb34714af58e45eb : SUCCESS in 24m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a5d1402d2da414a821110f14074a8cd : SUCCESS in 8m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87cb35ef423f4bd68ca102a12d5bd4de : SUCCESS in 19m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/af4eae88fd0f4d3fbf1b4a3c4b41e724 : SUCCESS in 47m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c988a1c980fe455d9ad98a2cd376faf1 : SUCCESS in 50m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e536067f9ef541688137e7b7156891a0 : SUCCESS in 44m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5994071938b4dc99113fa7a80f6d8cf : SUCCESS in 36m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79b29338b5494977863c7cebbc300157 : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4d0d7e1d4847c48258f4260414a251 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10a7e1367b1f412c8c908c46678aa1ab : SUCCESS in 26m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/383f247c57b144c8b15af32dfff6a13e : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"83aa7cc5beaa6f774b31510fd76f06ea63f662c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 06:28:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e9530ed548f8edd3a4473f1cd3499b4a68c3ab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 07:31:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8ccb855d3c4898ac2bc3ebcde03c36 : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/f8b9e5b087c44b99b5d821af2f2aaaf4 : SUCCESS in 44m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9fd631fdf5344001a3db26d2ee94a032 : SUCCESS in 45m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e4f9997bd2d4dda91525786303272fa : SUCCESS in 41m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b37a2e0de18f40438e5b598b3640e5a5 : SUCCESS in 8m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f2664627a88b4df495543585b6fb710d : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c55c7ceaf09e41b4b9eb6ca442d5a87d : SUCCESS in 17m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d07f9d9ab3d4faf92463e5351eb7ff8 : SUCCESS in 20m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52d400591e6b41b98aed3b4b527ecfff : SUCCESS in 19m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7bdd102ba051432ab1607a4b63e36593 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59e8c41c4455426e8cfe3aec725ac9a9 : SUCCESS in 21m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/730387d8e0024916b6ff8db78290c3cc : SUCCESS in 35m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d4a93e48c694c7f8e929621316e44b0 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9596ddc0e2b347a19683fed0769b5b49 : SUCCESS in 20m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7ad58c92ff864a8889ad66575f83e29f : SUCCESS in 29m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/256c7e13ab454c46aef623c43589a607 : SUCCESS in 5m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/34cae65741784cd7920563418dcca19d : SUCCESS in 44m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/45886d4b770741b69ffbd4adc8e73cde : SUCCESS in 1h 02m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de47309410aa46df9c91be985baec926 : SUCCESS in 35m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/742b8bc988e74bcfa1ee6ddd14254f7f : SUCCESS in 38m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3570fc0e45c465fae5a03fb9608673f : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aefbd1be8d64996aa993d368c7944b9 : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f92a9978abf4b62bcdc10dd189f0170 : SUCCESS in 21m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce39ca8dd4564a49a021685710e9d772 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1c01e0067880be30590ababf92282df3bcf819fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 07:31:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"594dbca196e2512ea94535a45cf43d820acf76e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 07:34:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bffd8df8c724b50af89b8d8204f7393 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7c9b398016974e0ab9b4156617bf3784 : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e571a86c6e9a47cc8e62e268733c4eb5 : SUCCESS in 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8773877e47378f07c781e66ff292282b02a75c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 07:53:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46061d0e71574196b5b901de6403421c : SUCCESS in 54m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97cd1a5b342b4f3f8c0a85440e2498cc : TIMED_OUT in 1h 31m 23s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/6aa222eae17a4952b6c063ec86768584 : SUCCESS in 2h 15m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fac1a29f32c4657bc718c2ba965a61c : SUCCESS in 1h 03m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c264f3260b464b5b96115c2b96774c7e : SUCCESS in 1h 05m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac92b3d8d83385eb0652e7894cc36b4f732f0174","revisions":{"ac92b3d8d83385eb0652e7894cc36b4f732f0174":{"kind":"REWORK","_number":1,"created":"2021-12-10 21:50:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/821478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/821478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/821478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/821478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/821478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/821478/1"}}},"commit":{"parents":[{"commit":"d6206e6fe5ad8c8c4f836ff7edd4f7e64acd3c17","subject":"CI: Move py37 and py38 jobs to experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6206e6fe5ad8c8c4f836ff7edd4f7e64acd3c17"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-10 21:50:29.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-10 21:50:29.000000000","tz":-480},"subject":"CI: Add rolling upgrade job coming from stable/xena","message":"CI: Add rolling upgrade job coming from stable/xena\n\nChange-Id: I17638d469f704e106ce691bed51e879c736ce153\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac92b3d8d83385eb0652e7894cc36b4f732f0174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac92b3d8d83385eb0652e7894cc36b4f732f0174"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
