)]}'
{"id":"openstack%2Fswift~821614","triplet_id":"openstack%2Fswift~master~I727fd96472bd96ce0318224470866d603f4c52d7","project":"openstack/swift","branch":"master","topic":"bucket-ratelimiter","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-21 18:25:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I727fd96472bd96ce0318224470866d603f4c52d7","subject":"Defer updates at most once","status":"ABANDONED","created":"2021-12-13 19:01:39.000000000","updated":"2021-12-21 18:25:58.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"03aa12fbc8f7fc3a0dc3340d2999303e7591af03","_number":821614,"virtual_id_number":821614,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-13 20:34:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 19:49:21.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-12-13 20:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e42efc52d51ddc2e210d25520679d802c7295c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-13 19:01:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a80999b3aa66c4469dda66d8bdc95f030966c2f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-13 19:49:21.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3fd205e700605517fee6a601de818627686aa396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 20:34:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f065814ca64e6abfc6d63ba0582a0c : SUCCESS in 10m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/390246ee7b3b46e48fdb4141f6a1972d : SUCCESS in 57m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20de66e8b1e34ad8a0a2de2ab368f7ee : SUCCESS in 47m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1d534f8934f49158f70aa3d6093eb5f : SUCCESS in 48m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/009864d21dc441f7867eb1709aa2aabb : SUCCESS in 7m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8bb2f1faed074436be5f4defe78219b8 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ecc3b4ecdada469eb417e75a186a4120 : FAILURE in 17m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52a6e2d682fa498087da2fd14ccca0f5 : FAILURE in 16m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e932a58970fb4952ae237fbe28e51b67 : TIMED_OUT in 1h 31m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e46a7a2bb9e047f5ab9bd75a01024054 : FAILURE in 13m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6bc44a8d804049bca2444e16068d0c0c : FAILURE in 19m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/116c62fc8fdc4ec5a18100edab61d6bf : SUCCESS in 21m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8aa6851f2ef24258b1373edda4d4974c : SUCCESS in 33m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2cbf04b0fc09406ca46fafc1cd6f37c9 : SUCCESS in 22m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c33de4a9fcaf46e9b23a5dced69cd116 : SUCCESS in 20m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/abdc335b65bc4635b29056b7dda6d0e4 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38f63f82d57343b9b851d153791bb77a : SUCCESS in 22m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fc7c0c2bd2c245babb127ae3699ed154 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7dfb3a747a654847b423f6246e8a25d7 : SUCCESS in 26m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e003cce486eb46d383c37daa9d92f668 : SUCCESS in 56m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1180e7d7fae0440fbd95632cce437dcb : SUCCESS in 52m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/481510b2220a4b6bba82ed4bb914e8aa : SUCCESS in 31m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2dcc6cd4ab349a78a510930d011d2c5 : SUCCESS in 40m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca9b3b7208af4dacaf734372547e2ec1 : FAILURE in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5080b6ed9953430291c1d1371a1ecbd0 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ac8235c78714a14bfc359c54929c760 : FAILURE in 12m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57f15050d3284931bf3444b2b5241625 : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1795032e2cc9c5ec21724c515af2000fc68c7e98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 20:40:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b54a062d7954d19a5d1ae9f8aa53395 : FAILURE in 31m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/030588069e1b49ac828b7b2c9e5dda43 : TIMED_OUT in 1h 31m 31s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/14a555065a16457b9fbf91d8d84a294e : SUCCESS in 1h 30m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8cf8dd0c3b44e02b86564b3cf511bf1 : SUCCESS in 33m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6eeb05542a2446db4af84a47834c1a7 : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":1},{"id":"03aa12fbc8f7fc3a0dc3340d2999303e7591af03","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-21 18:25:58.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/swift/+/821736/ is better","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"111e96342cf49413331e80936067773c54b27bb5","revisions":{"111e96342cf49413331e80936067773c54b27bb5":{"kind":"REWORK","_number":1,"created":"2021-12-13 19:01:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/14/821614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/821614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/821614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/821614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/821614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/821614/1"}}},"commit":{"parents":[{"commit":"6edac846f407f7297909e46d9ba9514281940e59","subject":"Deferer most recently parsed updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6edac846f407f7297909e46d9ba9514281940e59"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 19:00:33.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 19:00:33.000000000","tz":-480},"subject":"Defer updates at most once","message":"Defer updates at most once\n\nOnce we get into the deferred work, have the bucketizer start rate limiting.\n\nChange-Id: I727fd96472bd96ce0318224470866d603f4c52d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/111e96342cf49413331e80936067773c54b27bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/111e96342cf49413331e80936067773c54b27bb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
