)]}'
{"id":"openstack%2Fswift~821627","triplet_id":"openstack%2Fswift~master~I388420fa77358f7ad54076d10faf17f19d64ffbb","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-05 15:11:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I388420fa77358f7ad54076d10faf17f19d64ffbb","subject":"CI: Move rolling upgrade jobs to using CentOS 8 Stream","status":"ABANDONED","created":"2021-12-13 22:03:13.000000000","updated":"2025-05-05 15:11:39.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86a72519cc26a9c46664f200d691feee492c48f4","_number":821627,"virtual_id_number":821627,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-13 23:45:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-12-13 22:56:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-13 22:03:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-13 22:03:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2713b782c950c4dd7b7ce8390e869e2b12337d4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-13 22:03:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87453d3c41d42657a58def03f05e131c3bd36f7b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-13 22:03:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4b2e1a07017569a0f345f6058075618a6de6079","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 22:03:42.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch feature/losf.  The error was:\n\n  Nodeset swift-five-nodes does not match existing definition in branch master\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: swift-five-nodes\n      nodes:\n        - name: test-runner1\n          label: centos-7\n        - name: proxy1\n          label: centos-7\n        - name: account1\n          label: centos-7\n        - name: container1\n          label: centos-7\n        - name: object1\n          label: centos-7\n      groups:\n        - name: test-runner\n          nodes:\n            - test-runner1\n        - name: swift-cluster\n          nodes:\n            - proxy1\n            - account1\n            - container1\n            - object1\n        - name: proxy\n          nodes:\n            - proxy1\n        - name: account\n          nodes:\n            - account1\n        - name: container\n          nodes:\n            - container1\n        - name: object\n          nodes:\n            - object1\n        - name: storage\n          nodes:\n            - account1\n            - container1\n            - object1\n\n  in \"openstack/swift/.zuul.yaml@feature/losf\", line 276, column 3","accounts_in_message":[],"_revision_number":1},{"id":"cea22bcd63873408f044ca357c051097480bc5cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 22:03:44.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch feature/losf.  The error was:\n\n  Nodeset swift-five-nodes does not match existing definition in branch master\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: swift-five-nodes\n      nodes:\n        - name: test-runner1\n          label: centos-7\n        - name: proxy1\n          label: centos-7\n        - name: account1\n          label: centos-7\n        - name: container1\n          label: centos-7\n        - name: object1\n          label: centos-7\n      groups:\n        - name: test-runner\n          nodes:\n            - test-runner1\n        - name: swift-cluster\n          nodes:\n            - proxy1\n            - account1\n            - container1\n            - object1\n        - name: proxy\n          nodes:\n            - proxy1\n        - name: account\n          nodes:\n            - account1\n        - name: container\n          nodes:\n            - container1\n        - name: object\n          nodes:\n            - object1\n        - name: storage\n          nodes:\n            - account1\n            - container1\n            - object1\n\n  in \"openstack/swift/.zuul.yaml@feature/losf\", line 276, column 3","accounts_in_message":[],"_revision_number":1},{"id":"53c06f77f1c2948ade4c1652350a37710d9dbb7d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 22:04:06.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/swift on branch feature/losf.  The error was:\n\n  Nodeset swift-five-nodes does not match existing definition in branch master\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: swift-five-nodes\n      nodes:\n        - name: test-runner1\n          label: centos-7\n        - name: proxy1\n          label: centos-7\n        - name: account1\n          label: centos-7\n        - name: container1\n          label: centos-7\n        - name: object1\n          label: centos-7\n      groups:\n        - name: test-runner\n          nodes:\n            - test-runner1\n        - name: swift-cluster\n          nodes:\n            - proxy1\n            - account1\n            - container1\n            - object1\n        - name: proxy\n          nodes:\n            - proxy1\n        - name: account\n          nodes:\n            - account1\n        - name: container\n          nodes:\n            - container1\n        - name: object\n          nodes:\n            - object1\n        - name: storage\n          nodes:\n            - account1\n            - container1\n            - object1\n\n  in \"openstack/swift/.zuul.yaml@feature/losf\", line 276, column 3","accounts_in_message":[],"_revision_number":1},{"id":"cf76695d1e11a0e5c2dc6db4fd7a8da09913172a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-13 22:12:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50af0b0debfd231719a27be7880a41dd066483a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-13 22:13:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bb97ff89a42832c7ba8d03d30164be5dd1373758","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-13 22:56:48.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e614a69c7443ab892ef5b05b824ec5abda3f1050","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 23:07:11.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/28c42627b0b64f51ba9ba377dd1e7019 : FAILURE in 24m 05s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/c693b73b0cb04c829020bc6458ab0b84 : SUCCESS in 26m 04s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/cf6b6dac1c8e4b5b9d87b53b10d5abb4 : SUCCESS in 49m 58s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/4a0224eb0c8643f98a3dd699c175a4ac : SUCCESS in 28m 03s\n- swift-tox-py36-centos-8 https://zuul.opendev.org/t/openstack/build/e244220da3614541b49cdb6ba2eb4aff : FAILURE in 22m 26s\n- swift-tox-func-py36-centos-8 https://zuul.opendev.org/t/openstack/build/bf362ee7cbb44848b52c3d51e68293a1 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/5813d147fdd74b67a4b69dab3bef5389 : SUCCESS in 22m 44s\n- swift-tox-func-ec-py36-centos-8 https://zuul.opendev.org/t/openstack/build/459ac678780044379f7d7f4e482ed684 : SUCCESS in 30m 53s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/c113edc0ccf04762884ac16282ab542d : RETRY_LIMIT in 5m 20s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/79381a4816594dd595acb2a154de39a3 : RETRY_LIMIT in 5m 08s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/61ba27abf7514915ac2fcbb08ee22f5e : RETRY_LIMIT in 4m 43s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/ef7d1e940e1742238c21bf1b32dd4d82 : RETRY_LIMIT in 4m 53s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/7956cff55fb24fd0b43e1e8dfa814d8f : RETRY_LIMIT in 6m 02s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/f270ec36bce74daaae29867dff454705 : RETRY_LIMIT in 5m 59s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/3bf5abf57727497eb8429e7099e0f3bf : RETRY_LIMIT in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ff326859584c9fa1041c1419f31500b563e0c9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 23:45:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75e5cd8842247b8b11d57fbb12a4c03 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/20796c89da794a5d9f08c813637c5356 : SUCCESS in 53m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e18cb835f21346cd8fdf1cfea6085846 : SUCCESS in 32m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad7fae5a8313479ba250f70eedcb8d55 : SUCCESS in 45m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/504ae6dcc2d841148e5e0b027a76aa91 : SUCCESS in 7m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b2737818cccd42b3a3d247fbaa6be44c : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7078a232bfc0495782e417986a5e55d9 : FAILURE in 12m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f928b4e1ed8d4436b7937829a565a2d2 : FAILURE in 22m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/59da94ecef7048f99ba043bd96fff76e : FAILURE in 14m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fc12a64ae7274f06bae8b38b120baad7 : FAILURE in 18m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa95fe88c3db43928b6930df7b7e469a : TIMED_OUT in 1h 31m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/358a7e145cca49d6815a14378fec172b : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea62971c38a74966903cc203c4bd2930 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f74e9216b66c4f1d96ae55af46a44910 : SUCCESS in 23m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ac1c4b3a7533484080cadb9ee67640bc : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f38bf04f70c54102b6ea27d329089dd9 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/61a30ce8f5f642659aa64ba1ca5eaf08 : SUCCESS in 18m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d19c58544da741e8ae33067a93204a88 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc91fca790164b0687c3ca20df9603f9 : SUCCESS in 26m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2c12a76291fe42e2b0e3e22127ab876f : SUCCESS in 41m 54s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7f75a145e9754dcb915c47f402ec14e0 : SUCCESS in 55m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a4558aa88d8412282774c0bbe5de2fd : SUCCESS in 41m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/065e27bca29b4a60a52dcb47ed60c7fe : SUCCESS in 38m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/263486c9a78c43dba1f36079a9b302c6 : FAILURE in 15m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7b9c26e9e6b4ed7b87c803db90ba4f6 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d3d029172a5491c9c07f82bc67addc1 : RETRY_LIMIT in 5m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/144a59e2830a4ed89c178537b89c3455 : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":2},{"id":"0390ed33e937ad929ea20517d8c1bdcdd8198e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 23:55:26.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f59a6697ad3a44da81612ac822c2ec7d : FAILURE in 28m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d05b870bb58c4ef1911f8b2554dbe325 : FAILURE in 53m 49s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/eb93756572ea452dbf94d84a59b69374 : SUCCESS in 1h 36m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/02c36fab3cd649169ccb567e42661566 : SUCCESS in 55m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5a082ef5ed44f5d8cf013119abcba73 : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":2},{"id":"86a72519cc26a9c46664f200d691feee492c48f4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-05 15:11:39.000000000","message":"Abandoned\n\nYeah, we already moved off 8 to 9 :P","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e431ec174ff1551e8f914d125d504a2771833dee","revisions":{"95ef50bdce63cc1066fa56cd1b64812cc65b54fb":{"kind":"REWORK","_number":1,"created":"2021-12-13 22:03:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/821627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/821627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/821627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/821627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/821627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/821627/1"}}},"commit":{"parents":[{"commit":"1859f2e161f396023d8328489fb4b88c303797a9","subject":"Merge \"reconstructor: include partially reverted handoffs in handoffs_remaining\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1859f2e161f396023d8328489fb4b88c303797a9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 21:57:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 22:02:43.000000000","tz":-480},"subject":"CI: Move rolling upgrade jobs to using CentOS 8 Stream","message":"CI: Move rolling upgrade jobs to using CentOS 8 Stream\n\nChange-Id: I388420fa77358f7ad54076d10faf17f19d64ffbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ef50bdce63cc1066fa56cd1b64812cc65b54fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95ef50bdce63cc1066fa56cd1b64812cc65b54fb"}]},"branch":"refs/heads/master"},"e431ec174ff1551e8f914d125d504a2771833dee":{"kind":"REWORK","_number":2,"created":"2021-12-13 22:12:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/821627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/821627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/821627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/821627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/821627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/821627/2"}}},"commit":{"parents":[{"commit":"1859f2e161f396023d8328489fb4b88c303797a9","subject":"Merge \"reconstructor: include partially reverted handoffs in handoffs_remaining\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1859f2e161f396023d8328489fb4b88c303797a9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 21:57:06.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 22:12:14.000000000","tz":-480},"subject":"CI: Move rolling upgrade jobs to using CentOS 8 Stream","message":"CI: Move rolling upgrade jobs to using CentOS 8 Stream\n\nChange-Id: I388420fa77358f7ad54076d10faf17f19d64ffbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e431ec174ff1551e8f914d125d504a2771833dee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e431ec174ff1551e8f914d125d504a2771833dee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
