)]}'
{"id":"openstack%2Fswift~821758","triplet_id":"openstack%2Fswift~stable%2Fxena~I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","project":"openstack/swift","branch":"stable/xena","hashtags":[],"change_id":"I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","subject":"Fix cname_lookup test","status":"MERGED","created":"2021-12-14 23:48:48.000000000","updated":"2021-12-15 06:59:00.000000000","submitted":"2021-12-15 06:57:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821758","cherry_pick_of_change":821625,"cherry_pick_of_patch_set":1,"meta_rev_id":"d50eef8e6c4b593867683f472ccbfb444289bee9","_number":821758,"virtual_id_number":821758,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-15 06:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-15 06:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-15 06:11:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 00:44:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30c3db8df9cf82f56f3ac65b5cf58eef2d5f7aca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-14 23:48:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"797b7dc5560264936c917d5cfbdff94a182702c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 00:44:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7696d897c2ad4144ba0d43bde1ee804c : SUCCESS in 8m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6027bf33bb9f47b08ee04f74a82da623 : SUCCESS in 41m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a75f3c56dcf4821ba183626e97b1fdb : FAILURE in 6m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4d820464b6744f99b76f545a6dabd6c9 : FAILURE in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d207fabbac1b4cf3bef57b9920c96e32 : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e255e00f6c142a98655e9f2c089fa0d : SUCCESS in 16m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/936ee8810acc4a79a3cc0b2875eaf2b7 : SUCCESS in 15m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f020678637ce4706842b13d73ea4036b : SUCCESS in 13m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98b428b50da74c8da442d714c0277489 : SUCCESS in 21m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d7947074e1a4e56bc61f9e2f2514cd6 : SUCCESS in 23m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9939dafdc0a348e7952d8199fe343f88 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eba4b77ba4894bef909590fd2fbe7a97 : SUCCESS in 18m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a71d54a3740a4d69a4aef6fd80bb3ecf : SUCCESS in 31m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a1ac190461cd4de89d9144323d49d7da : SUCCESS in 25m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/edc5679045c44d88ac7e44f05323d570 : SUCCESS in 23m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15c522b6ecbb42ad996b655fd6143dd5 : SUCCESS in 52m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/589b1fa2d3c3464aadbcf33e4cded044 : SUCCESS in 38m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3102a09683f040d5aea3068cc6f51a4f : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73689bffd3c74958ac7af5c42f6d8055 : SUCCESS in 6m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b3f9a6f420b4b9e8894543bc8f3aba0 : SUCCESS in 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8c9f40beafa7d1287c5f392e6985350e3a007b46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 01:33:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a654bc824a144963927925a708507f35 : SUCCESS in 53m 27s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0da8de017e9e4feea3efdb214b2ec284 : SUCCESS in 55m 06s (non-voting)\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/a3a92dc7a06b466f951aaaee0d68aece : FAILURE in 1h 36m 42s (non-voting)\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4380594c6c2047d39a2e38b2f159f0a8 : SUCCESS in 51m 23s (non-voting)\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e4c2215026845f6b3dd27d0cefcb129 : SUCCESS in 45m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06b9b1de76fe73bbe9ad9744b23e66ba0d23b200","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-15 06:11:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c66e4287ba38033c80fd6fa66193d1ae96cf6c5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 06:12:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0734925a77467c44006084ff8c67f0cd279e7a82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 06:57:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ec2703b7d84b18adf18c954c2aff99 : SUCCESS in 9m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f327f6d19544d90a004a735cb66bd49 : SUCCESS in 44m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ebd5f2e482c340b1830808e5bc1c84b5 : POST_FAILURE in 14m 58s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bb23facf776d4094b4a56c5bf0161269 : POST_FAILURE in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bc0e14010494448fa759551813861d6a : SUCCESS in 16m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/abf86df400dc4f06ab2f20509f68a33e : SUCCESS in 16m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05f1e58e7c634accae0a6e7f89ff3ca3 : SUCCESS in 16m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2004a88fb170455b9b67c02f6f5d32d3 : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb5ffc3b730d4436aed03f2eb9e0cd82 : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2622fb4fefd14282aced624af99f1290 : SUCCESS in 23m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a9a65fbe244e4bd0921b66f733a2127a : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8c6d849dff94563ab624b35f92c4dc1 : SUCCESS in 28m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3981423d271f45cf99566c02b83c1b78 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d51ef8e2d94c40ff8f2222286fca9ac7 : SUCCESS in 16m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/53550b49fa24434f8b0bb0400889f55e : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a47c2d499a44176a273b0a518734cb2 : SUCCESS in 5m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c9f0516858e47f79696843e03c3ba78 : SUCCESS in 37m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5fab9dcc815543ff8d829e70adf38a68 : SUCCESS in 40m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69dcb0e612ec4174ae538193972fbb6f : SUCCESS in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0cd53110ba8464d87a7904a5b3cc246 : SUCCESS in 8m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f4d4eb22ff434389b2a5fcbfaf9268c0 : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7403ac20f8e400a36cba2274b7a4596e7092ccdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 06:57:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d50eef8e6c4b593867683f472ccbfb444289bee9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 06:59:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee50d5870f04acc8878c67709e37bc0 : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/60815d2ce8b14fd482962dba271b7167 : FAILURE in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/88be8ac417b44949a3acab0fd29e1b22 : FAILURE in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68bd795a6439baa869535944380996ca365ae4c2","revisions":{"68bd795a6439baa869535944380996ca365ae4c2":{"kind":"REWORK","_number":1,"created":"2021-12-14 23:48:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/58/821758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/58/821758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/821758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/821758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/58/821758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/58/821758/1"}}},"commit":{"parents":[{"commit":"8c41bf79b610ae21f4db15d9d118a54d570caf20","subject":"reconstructor: remove non-durable files on handoffs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c41bf79b610ae21f4db15d9d118a54d570caf20"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 20:58:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-12-14 23:48:48.000000000","tz":0},"subject":"Fix cname_lookup test","message":"Fix cname_lookup test\n\nc.badtest.com actually has a CNAME now, and apparently we\u0027re actually\ndoing the look-up during tests.\n\nChange-Id: I306b7d05740a2b8fcef2f5f432ebf5211bc723cc\n(cherry picked from commit 54fc8a7dee4ad7a38944cbd4c2e3b5f2ec393765)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68bd795a6439baa869535944380996ca365ae4c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68bd795a6439baa869535944380996ca365ae4c2"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
