)]}'
{"id":"openstack%2Fswift~821759","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","project":"openstack/swift","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-15 18:37:57.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","subject":"Fix cname_lookup test","status":"MERGED","created":"2021-12-14 23:49:08.000000000","updated":"2021-12-15 19:26:05.000000000","submitted":"2021-12-15 19:24:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821759","cherry_pick_of_change":821758,"cherry_pick_of_patch_set":1,"meta_rev_id":"2bdbf687ef5fbbf5d014834da52a1631638a60c3","_number":821759,"virtual_id_number":821759,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-15 19:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-12-15 10:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-15 18:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-15 18:37:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 00:43:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 10:54:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"f93c65d349065fc25b6a84156898b7d9a3f9a6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-14 23:49:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"2ee2addd8f1b83aacb76baa904ae3bfe7bc638a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 00:43:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8be47dcd5304aaea1e0d56c57bdce6a : SUCCESS in 7m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c0e3a9295f14fd2bac7dbc1659f1d6a : SUCCESS in 50m 57s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2c672b1a8a94b438d228c4fb171eacc : FAILURE in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/79b1c67e97f94c9e9f1ef3360eb4b1db : FAILURE in 6m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c7a8e510f3b425ab06a993852e5801e : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f95bb41c85b4474b0d603dfcf29b789 : SUCCESS in 14m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/85e2c49e1f6d48d3b1c6e6dae4bae2a2 : SUCCESS in 14m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5679d379c8ea4d22b4de391ce2ccdf20 : SUCCESS in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3a270b7203a64076882449fccbbb372a : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/831579ab8425469682f3eb5af48d2f90 : SUCCESS in 21m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da1d3186648d4e098e9f99da25e1cf23 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/44e1c3a08a5e489ba342753707f5d0fa : SUCCESS in 21m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cf6f96dd6513434d9d5609afc246db63 : SUCCESS in 22m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6ba02521eb174e2abeeb448fa0c14e9a : SUCCESS in 23m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/49b91f13bd96452d8d56b4afd52cd62a : SUCCESS in 18m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd34db9a19d94dc188d62789d7be44a5 : SUCCESS in 28m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3aa98e4bea84444084c95e60f84c4611 : SUCCESS in 29m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d844bb8ab203479e89eeab45a1004c6e : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0428f3c5da624d51bd21e802b8a2f0be : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21014dfe1ddf44ef987a3212f4850f84 : FAILURE in 26m 05s","accounts_in_message":[],"_revision_number":1},{"id":"13dfbaf3e62b018fd8a6818a6e06a13e00c6da90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-15 06:12:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8892d04b8ec69d203e54c413e479190007241355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 07:21:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/818a544c2ab048f099a28351798a2f45 : SUCCESS in 8m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2708aa8926c4137b9f90776c813320c : SUCCESS in 1h 02m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/354bcf36070f4b348f2ab9da05def028 : FAILURE in 7m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f6155c9d39894e33927708a2cc7fd450 : FAILURE in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c50b0c862cc4d738d7b52c980d63883 : SUCCESS in 18m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3c25b9164441c5a1f59bedef4f6e3f : SUCCESS in 13m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/946baa4beb80490db92bd2e7a98b104d : SUCCESS in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7e7f4d38ce05407e99d7326a50b0a13d : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/68dcc6061f784ec6a87ba4f75fb898dd : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e7f9e7eeb59b407da3169354b3086cf0 : SUCCESS in 23m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa8cb7fe20a24769b0bfc9c70e678ac5 : SUCCESS in 25m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/23d36a00bc4d409cbe384d1a64ced218 : SUCCESS in 16m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cb66085fac1149c4bc17c36fdc761e32 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c8cb59ef633f4e6782f29583caa90327 : SUCCESS in 23m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/afc1268a63b348aa970c168b2a5c04b1 : SUCCESS in 30m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8834d67f61a4630b03b389deaf94f1f : SUCCESS in 40m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f56912bcbfb4bdab1ff3bd09da926ab : SUCCESS in 39m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5048046be3fb4b8ab04c32da4131db6d : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf28f0e8f2884691ba86278fb888644b : SUCCESS in 6m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2968578d034a42f08466bfff25d2cc5b : SUCCESS in 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"c229ff205cff7cd96483815f9ab602ebf4ffc231","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-12-15 10:54:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4896f8d5aeb1e3d96af01be34181682045c2ed42","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-15 18:37:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6c75698393e55397f090666e252b6842e60b7234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 18:38:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"463923691aff34b2a3e49948429b1a379f7675b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 19:24:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d92499079194a41a0d90a88d91a6bd4 : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74d1ec2037c44cb681ead1e8ffdc05a8 : SUCCESS in 42m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a7ac01523bb5487aa4fff290478fabcb : POST_FAILURE in 10m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/47e8aeaf74a44f52816b6ff6a1dca0f4 : POST_FAILURE in 7m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b58e600b0bc47a0a4e1604e4c987c62 : SUCCESS in 13m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91bb10daf6d542d69ddd872decfbc933 : SUCCESS in 17m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/267d5b803d134b7b87c0f0d69af8d206 : SUCCESS in 20m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/595d9700909b496da292cb2f334d31c1 : SUCCESS in 15m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da44c510788b4bf9a888328b7c8906cb : SUCCESS in 14m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b7df948b24c841ea947a39d221736c4f : SUCCESS in 20m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22e19a6259254c1aad3826fa54fa4a66 : SUCCESS in 25m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c759539628584c3e8c234fab817403f6 : SUCCESS in 26m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8714b82f50f44a9a801b58f1b5aa83ce : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/26c2c0b7e64c4a4981b212010f30973c : SUCCESS in 17m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2c003e91eac74508b7ea6bef708cc9c8 : SUCCESS in 31m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6db3dd728464498bc736f10ec191b40 : SUCCESS in 7m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8665342a74e446b095a9f81a650cb800 : SUCCESS in 39m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/792a18ab031940fe9e9bf774d4bcbdfe : SUCCESS in 38m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16e47472a4714cdc88515be2e6e8e65d : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aec70bfdd384f35a547ff9e7d4e25f3 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6999fba65f9459bbe0650ba06e28ca7 : SUCCESS in 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"527166148b2a1968127eb8130ed76d17b8db0dbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 19:24:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2bdbf687ef5fbbf5d014834da52a1631638a60c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 19:26:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3caf3d5879914d0b9f149e89a5e86f1d : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/639d18ce59d24c7db820c0f64aed6784 : FAILURE in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/09a56c07180744ef8e3c8618ec037b1a : FAILURE in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a25a936c4fd4486d66c434715f0cb8a6f404885d","revisions":{"a25a936c4fd4486d66c434715f0cb8a6f404885d":{"kind":"REWORK","_number":1,"created":"2021-12-14 23:49:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/59/821759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/821759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/821759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/821759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/821759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/821759/1"}}},"commit":{"parents":[{"commit":"c966628978ac4ceb1d94ba92879df071b8ca265d","subject":"Quote paths before sending them to swob.Request.blank","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c966628978ac4ceb1d94ba92879df071b8ca265d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 20:58:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-12-14 23:49:08.000000000","tz":0},"subject":"Fix cname_lookup test","message":"Fix cname_lookup test\n\nc.badtest.com actually has a CNAME now, and apparently we\u0027re actually\ndoing the look-up during tests.\n\nChange-Id: I306b7d05740a2b8fcef2f5f432ebf5211bc723cc\n(cherry picked from commit 54fc8a7dee4ad7a38944cbd4c2e3b5f2ec393765)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a25a936c4fd4486d66c434715f0cb8a6f404885d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a25a936c4fd4486d66c434715f0cb8a6f404885d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
