)]}'
{"id":"openstack%2Fswift~821760","triplet_id":"openstack%2Fswift~stable%2Fvictoria~I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","project":"openstack/swift","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-16 16:39:59.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","subject":"Fix cname_lookup test","status":"MERGED","created":"2021-12-14 23:49:23.000000000","updated":"2021-12-16 17:45:00.000000000","submitted":"2021-12-16 17:42:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821760","cherry_pick_of_change":821759,"cherry_pick_of_patch_set":1,"meta_rev_id":"05631cb5c0e636538c646c82de2b4d365d1abb75","_number":821760,"virtual_id_number":821760,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 17:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-12-16 14:47:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-12-15 21:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-12-16 11:39:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-16 16:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-16 16:39:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 00:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 21:58:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-12-16 11:39:17.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2021-12-16 14:47:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"823e8953e7f0bc208d9c0ad889b606246ea4ee00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-14 23:49:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"ff47b15777be53b5e52a29b065b0e9c32262f41f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 00:49:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc517b0216844e83a3f0b923d9dd1eea : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0738dfe55db049f3908da9f0c656079a : SUCCESS in 52m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dce09834ed4d4c4980e2167485620092 : FAILURE in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c3bff504d9c747119e94bbb247117e7c : FAILURE in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/107d1e33cbc54dc4acf3725ced3eba5a : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ac0b6900c8e4d60909515eaa9b088b8 : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/019a83ee627444e9a9daaf3940cad71a : SUCCESS in 17m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/219459cfde254329a9343127c68c6495 : SUCCESS in 14m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34d5f3dbb9a24ba686007cce11125e9f : SUCCESS in 22m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae138c6f67bb4931a01af649184214e9 : SUCCESS in 26m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da06d5bda4fc43dfa9c07623d508d670 : FAILURE in 30m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2de1a4a63ad04339a482a31eb0ac4c79 : SUCCESS in 21m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a46495619fdb4148afbb537fa2e9a850 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8cdd82f164ec4199b28bec63c99f0f5c : SUCCESS in 19m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58cca4d1a9bc4e5ca2d5c04b7bb5b75a : SUCCESS in 55m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/586f5ccda7ee4cc3861ff4195b95acc9 : SUCCESS in 29m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19f7112b922d42bb870b4ee63373e7ab : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37592233dcf741498f32e48453c3d90c : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6952284fa9af49188ecbf9ce74206ac2 : FAILURE in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0207cfebc6ebb7890ee8a2630016fdec4841b58e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-15 06:13:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8e355b749e51b7eb7af1a9c0734ae281d5e1e1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 07:21:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3417676dc7514e71adbd28e14afe075a : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42510e62e24545cd9700f80171ffe49b : SUCCESS in 53m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bfa99dfcab674619a7c8e0fec2cfaacf : FAILURE in 7m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c2a13cb054904f5fbacd34daf09d23c3 : FAILURE in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c3fc077d943649d4af68cceca1492906 : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b53c317a947b43438f7aa2d87855af41 : SUCCESS in 18m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8319976a336048fe991e3f6ff1d8fe20 : SUCCESS in 14m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bb5415f769004d759e522d8b64bef479 : SUCCESS in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23e16b4102e140129d5cdba694967b74 : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1bdd27e7b22f43a7a2e5c98901933f88 : SUCCESS in 23m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a96090bcc6c343809bbf2025cfd95135 : SUCCESS in 19m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bc2b94e3f8e74411a33c18a4e5a72097 : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cbd8cc33cfa641d5892127a7346c4ccb : SUCCESS in 20m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3332473bfd174ad096eeb91c1d5b6f59 : SUCCESS in 22m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/20c1936479ab4c7ab4362d84979ed3fe : TIMED_OUT in 1h 06m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9245a3eda00547f4a07e2f0fab481e46 : SUCCESS in 46m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef97b7b8bf1940289a482df3118803f6 : SUCCESS in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3adaa9458cf846cb8db9752aaa21ca50 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eef98a14d86649d89c8977ef93a70788 : FAILURE in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bef90b790d0a4f3ef332151ca5615c8d0c05644","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-15 18:39:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"848ff74695dcce36a1aab8a31f2e7a4514d877ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 19:46:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b43544c9f1624a52a6271eeb4211497a : SUCCESS in 8m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e92774734ef24dfd9318855f9636c6b2 : SUCCESS in 1h 01m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e0369ff43cb4be3b9af6ac0280568fd : FAILURE in 8m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4cf5060548dd4114b9109a5645b1fcaf : FAILURE in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/462fb312dbbe4cb397a0e8793ea3ea17 : SUCCESS in 13m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5adbf0a3585f4d45b4e9426fc8914119 : SUCCESS in 13m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d084c09ed264340b78b4b34b4eda157 : SUCCESS in 14m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7663d3217984d408f5630f891a8332b : SUCCESS in 13m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99d2e42b22424cda80584ac13af2d1dc : SUCCESS in 27m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eb683ff58e21412ba65efa23b49e1646 : SUCCESS in 24m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f94e38002fa41ed886437e8d809f964 : POST_FAILURE in 19m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/64c1a0c407ff4c4293d35b2f22a769fd : SUCCESS in 30m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/048235cdb8b344518a43d9518acdfb45 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d37fc2b1f9204137a3c1efa1854f8c82 : SUCCESS in 17m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e105546f11444bf68d4dc9c9a67f8162 : SUCCESS in 47m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8dc1ab9e445a4876bea6198e26d29087 : SUCCESS in 47m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89ef6050d66d4b589ca27eda52aea24e : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa2241418ee4ab29c18b102fc5560e7 : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d65930c93ad4a9aa8efa43c06b77be3 : FAILURE in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"192b945ecedc59ae19c334072c27de2d2175f2dc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-12-15 21:58:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce7aa258a40fa7df3b1cd13bf93889366549b94b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-12-15 22:00:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a6ceb6d8557851baebf1f8f7af4de9b2e0f37a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 22:59:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a7d02664264981a192759bc1ea98b6 : SUCCESS in 8m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3458bcc4cf664d0b8bc21963ea8e8863 : SUCCESS in 53m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1102f52968a47d9bb4d13d668a8ef27 : FAILURE in 7m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/46b7e3734d37403581faa54fc5cd92a4 : FAILURE in 4m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/19881c57eb90455cb8c28c3f84125386 : SUCCESS in 14m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b9c45f4de55476a9fcb5369334356e0 : SUCCESS in 13m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/81583ea4c9c24a2c8c26054b3ed58a2a : SUCCESS in 17m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68e93da9196e4d3eaf29ef3d72c15392 : SUCCESS in 14m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/426c13fa697246e3b2c4ba5dee930069 : SUCCESS in 23m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/764724b6aed64bfdb21eb44dace2ed7b : SUCCESS in 29m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0865d8eabb1c46759b167ca296d372af : SUCCESS in 23m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3fce533657df498ebba036e61e1c0e8e : SUCCESS in 27m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8a3c139588634dc3a7e7151b7daf3ffa : SUCCESS in 27m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/55dad88acfed4612811146a9d384a39e : SUCCESS in 25m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/777bd7eb384e409494b81b351b47d94c : SUCCESS in 38m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ce81ae6dc184f86abe34ede20f697a5 : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83b1f6b1fb71492caca495b764044da0 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740d2eb0ac1443dfbb03a430dec51c27 : SUCCESS in 6m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c522160dbc84a78992e34e1cd838c73 : FAILURE in 9m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03ca73d7773ab2c3c6f48e1ce278bc905b99aa49","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2021-12-16 11:39:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7bbac94d61c5c2b2adc00ef5b98a1861c233d092","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 14:47:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"022dd961b406cf708ec7b38a0701cb3cb9257ee5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-16 16:39:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b28018e6c79f68d17cc41844974daeb47a020fd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:40:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"239bcc3b05cdb869aaad0ec87a9539e6a1a6ff44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:42:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0989772af8564565ae592099b0d289bf : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90853992cff34f80819536abf11a6942 : SUCCESS in 52m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f4d8ab6fd3704698a3c6407dfb5fef13 : POST_FAILURE in 15m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/95f5788238a34f359bb22445a9e4cf8c : POST_FAILURE in 10m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a146d6b837d7437fb32638ff33b20a55 : SUCCESS in 19m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9bce6e7dc439423fbea975990b658112 : SUCCESS in 18m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3088ecc68211458592c9e519ca6b3428 : SUCCESS in 14m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f8e703e9f1d842588e80b1783f5e3cc1 : SUCCESS in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dd39c8d7fae34408aa8f03d01726ef3d : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b69775b335e4c3bbb8308981b44cfa3 : SUCCESS in 26m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1304acda76cf4a04b4095ed0285b1965 : SUCCESS in 26m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8be4446f9f1b40e68d7a6e883427fc5b : SUCCESS in 28m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/88cdf42570c84020b204eff302986ea5 : SUCCESS in 23m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4682c38d5c044d38ffc676dbe11df96 : SUCCESS in 47m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97feca9d8ea848718c1a3ca416852f2a : SUCCESS in 48m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1f626ac20894f92b8014e0d2a4b060e : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46021842704c46578d1a593840193c09 : SUCCESS in 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0d4257f18baa05141905b56fffa6722fb4117729","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:42:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05631cb5c0e636538c646c82de2b4d365d1abb75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:45:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d841e2f0114ce8ade4b282080ff5e1 : SUCCESS in 1m 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/504027f194984edc953cae5bf0067e18 : FAILURE in 1m 05s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/66a0847496134056ba9d11db0b26dca9 : FAILURE in 1m 01s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f53500522ae52df3c07bdcf775d11b9adccc817c","revisions":{"f53500522ae52df3c07bdcf775d11b9adccc817c":{"kind":"REWORK","_number":1,"created":"2021-12-14 23:49:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/821760/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/821760/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/821760/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/821760/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/821760/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/821760/1"}}},"commit":{"parents":[{"commit":"41335e63ffd2b6c6262b70f8ef244737d18ab7ac","subject":"Merge \"symlink: Ensure headers are returned as a list on py3\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41335e63ffd2b6c6262b70f8ef244737d18ab7ac"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 20:58:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-12-14 23:49:23.000000000","tz":0},"subject":"Fix cname_lookup test","message":"Fix cname_lookup test\n\nc.badtest.com actually has a CNAME now, and apparently we\u0027re actually\ndoing the look-up during tests.\n\nChange-Id: I306b7d05740a2b8fcef2f5f432ebf5211bc723cc\n(cherry picked from commit 54fc8a7dee4ad7a38944cbd4c2e3b5f2ec393765)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f53500522ae52df3c07bdcf775d11b9adccc817c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f53500522ae52df3c07bdcf775d11b9adccc817c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
