)]}'
{"id":"openstack%2Fswift~821761","triplet_id":"openstack%2Fswift~stable%2Fussuri~I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","project":"openstack/swift","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-16 20:33:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","subject":"Fix cname_lookup test","status":"MERGED","created":"2021-12-14 23:49:40.000000000","updated":"2021-12-16 20:34:54.000000000","submitted":"2021-12-16 20:33:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821761","cherry_pick_of_change":821760,"cherry_pick_of_patch_set":1,"meta_rev_id":"010d67ddbd8a0a587b30c550ef1d377273995c41","_number":821761,"virtual_id_number":821761,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 20:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-12-16 14:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-16 16:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-16 16:41:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 00:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 14:47:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-12-16 18:41:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"9c76cec358e1341a32c63f8bf8f7ea0063f18f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-14 23:49:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"c98cddedfe36a3b048c0cc5c1634024733718024","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 00:35:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db76538f4b440cc860330d427dbf870 : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7af59855254c4e7984aaca0b40246265 : SUCCESS in 43m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97a033b9a2b74055840b67aa44ddd3be : FAILURE in 7m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cb946cb95bd14e50ab061de699d72e08 : FAILURE in 4m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/050a55433cad421c8a58e001f3f458d8 : SUCCESS in 11m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a7e05d352b8d43ebbf77ebc6cd8039ef : SUCCESS in 13m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b1f26133b4364ce5b0cf6a088a5f03dd : SUCCESS in 12m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c947fe3a30414734b5f7342a5e749c72 : SUCCESS in 13m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1111d07abf2c4921bb0247ef3887fcd7 : SUCCESS in 8m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5a8a4e3b0af4f98bece03cf18cd6d47 : SUCCESS in 12m 36s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/7af23f37c2364920bb1838d1ec73552d : SUCCESS in 12m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0b0af7f1086e4503b8ab213f15070206 : SUCCESS in 10m 07s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/feca6f39e68d4312b146326f9d9494ec : SUCCESS in 12m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e84b21ad53514d4491f19aa0355dcb07 : SUCCESS in 10m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/be6fbe69d34e42f787d8ed9ea49b0192 : SUCCESS in 11m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bd0e0c86ebc94ec198253ffb96c30d8b : SUCCESS in 11m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/59e64622ba4844c88602a5da737ce309 : SUCCESS in 13m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4da95c910db940279a841e0e4210ce44 : SUCCESS in 15m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f97fa56efbcb4093b2869c93c575690b : SUCCESS in 35m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/482659a9285d4bb1b64c2bb850aab80b : SUCCESS in 28m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2a6831a67e9e4dc2bd747b341f43b100 : SUCCESS in 29m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a77cbe36ff249c68a9d601ae941d478 : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b559e7024ab44b68315a003d4b5ceb4 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16b6636eb6ac4359a7c7d39a5308ffe6 : FAILURE in 7m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a2782c8fb85e1ac04aed840c3e21d356edc38db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 14:47:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7a4a17f8686838afd9062bca6f468103cc10888","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-16 16:41:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea046a2a7b055b40d5cece43536e0215a6d811b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:42:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"693f03093a2c93cd5d85b50126a7cb53ace0cd5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 17:43:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a412104d53b4b9293b6d6bd9420d2bf : SUCCESS in 9m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/678f1f9a0f614caead4186b46902a24c : SUCCESS in 56m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b6ffe4f9e2f6401396d025f08dae7a48 : POST_FAILURE in 11m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b4c583bacb7d4478a51111c9729a82d8 : POST_FAILURE in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/729dc0253e824552ac2e61e04314a5f4 : SUCCESS in 14m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a06af92eebc440a1b6ecdb88f43bae72 : SUCCESS in 15m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/887dfd6a91e04af394e42221e58d7a1e : SUCCESS in 23m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a177d18a666d4b96a0a4170323f6fe6b : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a1c6ad28e0854d8e848db26878df5965 : SUCCESS in 12m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/140a12f969e14810b4b355816ce0bea9 : SUCCESS in 14m 23s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/3fdcd5bcea1342a085b99ad96bd2f4b8 : SUCCESS in 11m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ed142b35ffb4515bc0ca7b24243c636 : SUCCESS in 14m 37s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ade8f27dbc2049ffb1a9d0b71456483b : FAILURE in 24m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b65c4783c1c24168a855c809b15033b8 : SUCCESS in 14m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/874c9395b5494f4fa18d414279eef6f6 : SUCCESS in 24m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/092c17e8af7f4de89113200ff631403e : SUCCESS in 12m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e875856407df46f79e89974b0b061808 : SUCCESS in 19m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2feb942aced4d5bae1fa7d019a35bc3 : SUCCESS in 24m 34s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9f25bd00a84444b88a945ef0d424f2d6 : SUCCESS in 34m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d2d3261acce443baed9568449e5d8a4 : SUCCESS in 34m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d4a71b2d50a4182b2bc2d0632ceebef : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32d4fd7c6cf34510b23e1dcd891065c9 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c050bfe24527f180c916227d4426b600fd58b32f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-12-16 18:41:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2da0a88c0ccb025b552ee057a1ce90cc63ae2b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 19:46:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8697931963f4f62a3ffb5d7c431972e : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af46876d06ce49b98644321af4bf530f : SUCCESS in 1h 02m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2db88e8fe31a4338973756e4345856b9 : FAILURE in 7m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9cc0594f2e014a56989da0adf7bdf4ba : FAILURE in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9ff2a8851df43b3b5c42735758e5e94 : SUCCESS in 11m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/826573b5aa8e4bd2991bc4f1406f820d : SUCCESS in 15m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/facffdab15a34449858f2134b107a641 : SUCCESS in 17m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b44360f206f4159a4aa2f3cacba965c : SUCCESS in 15m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e58a816c01d497998d43bdfadcf0c30 : SUCCESS in 11m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef95c0ab07594e7196353383f74104b3 : SUCCESS in 22m 39s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/a93b499b747949ac8741b63fa7d07703 : SUCCESS in 14m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1cfdb49b0e8f4c71b97c14e8f9bc2776 : SUCCESS in 20m 08s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/4a18a80443d84a4480f57efaa15e6569 : SUCCESS in 20m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b20c8cc822894fc0acfdf66285e98e73 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/32015e5679524f79b39bbeef00d6a3c8 : SUCCESS in 16m 38s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7e36d2a82b684725b0217780a15d30f6 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f5362d6d52b34082927168234839c8bf : SUCCESS in 14m 42s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3997bc3d3083456db70930cc2dac7ede : SUCCESS in 16m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0067f32a66734ffc917c93107c23d204 : SUCCESS in 27m 19s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/389b3bc5d39c46a9a8c64a4c870c4244 : SUCCESS in 59m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ede0560d8aa4428a3cb78ef26838b79 : SUCCESS in 22m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/840224986c094636896a19a4bf66bad0 : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e9cb6fa5b954b46a9e35b98ad844248 : SUCCESS in 6m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a567bce57b8b4598a74ef2d25cf8a399 : FAILURE in 9m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f948df3f6517219be501364addee23736ffa4a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 19:47:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7619bedbc873c7b15f6057ca34f9172ac34b7fa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:33:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba42e1bfdc24389991b22b06610ccb2 : SUCCESS in 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bf6840936ba426ca42f110594e14140 : SUCCESS in 43m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/00326cedf4454b7f8d806fdf64dee7c4 : POST_FAILURE in 9m 18s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cc43cf051a9e4b59bec30b1e6f68c6d9 : POST_FAILURE in 7m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd89ca71d00d4cbf9e11cdef597c4a7a : SUCCESS in 14m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01409cf2278c4382a3b7651043be2d24 : SUCCESS in 13m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a6767664ee434c7ba5a247e1a5206e04 : SUCCESS in 16m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42fcc06c67404329a56c1c02982d2663 : SUCCESS in 14m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d1e1a9997e84ec5944de8bd2f6a4047 : SUCCESS in 11m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0063a5f5db204732ada8f3568d142d06 : SUCCESS in 15m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/e9c16b0f7f204e4ca3fabac518a642a7 : SUCCESS in 11m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3519cc23bc3a4aaf93b7b672880ad776 : SUCCESS in 12m 13s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/ddd97d32d30b4872b6a7d1dbca632037 : SUCCESS in 23m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1325719286474cf48d88266e6eb85d44 : SUCCESS in 10m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9d1632c8cbbc4ebe82cd8f1f0aa1fd1d : SUCCESS in 13m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/77adbc14023640b7ba6f18c33b86013a : SUCCESS in 13m 31s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1406bf78a29d4a4c91dbebfe63ab0187 : SUCCESS in 15m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/314888dceb89490ab6e880665f4661b9 : SUCCESS in 30m 27s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/815e7ec4552c4162ab2931e6b8f383af : SUCCESS in 32m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a975bceeb961452c8880902fc07c685c : SUCCESS in 31m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bbf8790b2c440b3916d2a6bb9a68708 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6981217d6d04de5937e04a57f713548 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8905a8ec6a429f3139c6670c0ed7dab326fbfa77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:33:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"010d67ddbd8a0a587b30c550ef1d377273995c41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 20:34:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424d7a72ec5d401d827c63f92c5a41b4 : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a4172a0992c744c2b5cc91910acff7f4 : FAILURE in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6c5a5e1c550c4323bfb309193c961633 : FAILURE in 50s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ade6b52f5058fb5f99f77645a468841abb3fc892","revisions":{"ade6b52f5058fb5f99f77645a468841abb3fc892":{"kind":"REWORK","_number":1,"created":"2021-12-14 23:49:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/61/821761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/821761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/821761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/821761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/821761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/821761/1"}}},"commit":{"parents":[{"commit":"4b0d6a87e9b2d0a032fe3b3a85e6fef0e5e8839f","subject":"Merge \"symlink: Ensure headers are returned as a list on py3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b0d6a87e9b2d0a032fe3b3a85e6fef0e5e8839f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 20:58:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-12-14 23:49:40.000000000","tz":0},"subject":"Fix cname_lookup test","message":"Fix cname_lookup test\n\nc.badtest.com actually has a CNAME now, and apparently we\u0027re actually\ndoing the look-up during tests.\n\nChange-Id: I306b7d05740a2b8fcef2f5f432ebf5211bc723cc\n(cherry picked from commit 54fc8a7dee4ad7a38944cbd4c2e3b5f2ec393765)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ade6b52f5058fb5f99f77645a468841abb3fc892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ade6b52f5058fb5f99f77645a468841abb3fc892"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
