)]}'
{"id":"openstack%2Fswift~821762","triplet_id":"openstack%2Fswift~stable%2Ftrain~I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","project":"openstack/swift","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2021-12-16 16:42:11.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I306b7d05740a2b8fcef2f5f432ebf5211bc723cc","subject":"Fix cname_lookup test","status":"MERGED","created":"2021-12-14 23:49:52.000000000","updated":"2021-12-16 18:04:44.000000000","submitted":"2021-12-16 18:02:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821762","cherry_pick_of_change":821761,"cherry_pick_of_patch_set":1,"meta_rev_id":"f20a579918eb207acd57cbbcb988c252f81bc51a","_number":821762,"virtual_id_number":821762,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-16 18:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-12-16 14:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-12-16 16:42:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-12-16 16:42:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 00:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-16 14:47:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"02a9ea7e27bec0b648f9f2fd728cf55eaf44f221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-14 23:49:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"b2ffbe242e54992da1d55a49416b0caee7bea537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 00:54:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2cb32d461414485980b620c69555112 : SUCCESS in 5m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/78849766e7e24fc3afbb7a510cc44c89 : SUCCESS in 1h 01m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eae4fd94cbf4b57b826fa88a4e9ae6e : SUCCESS in 43m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c4f209d18ad44c9a64e9d4ea8d1d025 : FAILURE in 10m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f860cd751fd34c9ea8045d03c28a5af1 : SUCCESS in 9m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/05b78e8ab2ff4de4817eb2bc4c95ca5d : SUCCESS in 22m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4403688b16604a91bad41d7147d2a7dd : SUCCESS in 14m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1db7763c883540e59913d7b9c109ae4f : SUCCESS in 9m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/edd660d23471412cac83f803cb2aa114 : SUCCESS in 11m 12s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d55cfe95322641f3ab4759a86ceb60e7 : SUCCESS in 9m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17cadef5a6b149a1aa4072e3164b1c0b : SUCCESS in 9m 20s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/f0ba1865ceaf4f20b7d64401e7a6b0eb : SUCCESS in 23m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/714fbcf8919d4158a9cabed1d7b18f98 : SUCCESS in 9m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/65d7c35596264ec5986b2e501c041746 : SUCCESS in 15m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/addc7e2add824ba99677e81bbd0fa1e9 : SUCCESS in 12m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4e03206693ed432cb7aee034a45814df : SUCCESS in 10m 40s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4ee689483cab49e49534c61ae92a1150 : SUCCESS in 19m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1250c62d228c4fd39548c84f7b76414b : SUCCESS in 32m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/992287673095458484a24ade3203a8fe : SUCCESS in 28m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/591a140d399d441d906c4a48ecb2daee : SUCCESS in 20m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0e11df620564ebfb35ef785275cc9db : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57493f6847b242b8a0f6a14a00f1ffe0 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7a3ccd1cfc841ddb0c959a3590ec10a : FAILURE in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc8f8d6f9afd7322b5817f3ce1d3721652fd24a6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 14:47:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fbbc69cf10238d25bfe1e9a1c59029f3c262b1a4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-12-16 14:47:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89ba189c57e68e5d0eb887d7f73d3cc5f0b822c9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-16 16:42:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ad0a49dbc889331ec814aa069f12c3f9115eb3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 16:43:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9080151e1441f9741737a2ade4ec2276ff44744c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 18:02:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e8d567c0ccd41a3a53d2db61145d0db : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/77712fafb8044a929fcf283785857b58 : SUCCESS in 54m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd1d085dd65d45e192337007c83f8380 : SUCCESS in 50m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e5e3e17c489542cd983766f3f28a095e : POST_FAILURE in 9m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6767225a6acc48f59c9fec83fac67ce4 : SUCCESS in 13m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a236903239d749de9ede19ebb50f8b5d : SUCCESS in 14m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/47f48938f3364f55aba5d093107c94f2 : SUCCESS in 15m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e64868e2ba343d783ee3aa6ede7c486 : SUCCESS in 11m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d3ec3efbd63e4d17b6be6c1bfccac901 : SUCCESS in 11m 12s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/347c0b4a675d4a91905a0f538afa027e : SUCCESS in 10m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d827f5a1c650414b923281cfc3ff3b8f : SUCCESS in 26m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/03a550e6c70b49fcb20d6b9343aa3d2c : SUCCESS in 16m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/01c35ae58f55419db4b570d2931ef07f : SUCCESS in 10m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fb956fecf2a04572828f999ee4fea260 : SUCCESS in 11m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/165454d211b7490d98e32addf7ce6dcd : SUCCESS in 12m 47s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b4b9d1f74ce64f37868749a7520e492f : SUCCESS in 13m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75dbe38913f64146a6d3945beae41ad9 : SUCCESS in 34m 16s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/827cd3ee03614550b8ea741cb4e9795e : SUCCESS in 22m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2db46a1319a040e58f2bc0ffe27ea357 : SUCCESS in 35m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de070b6b1c9b4c2b843440d906c2a7a1 : SUCCESS in 14m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/630f5e0252174818bcb297011aae8924 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":1},{"id":"613a7a6354bf6c33e8e64d21fcf42e4154a7e428","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 18:02:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f20a579918eb207acd57cbbcb988c252f81bc51a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 18:04:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e730410f0d42828372e621bf924145 : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bc3b0071c6a844039c4a2b3fc72f9db6 : FAILURE in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bffa1e205c499768abd767f0ec4d6f0dc745e85","revisions":{"2bffa1e205c499768abd767f0ec4d6f0dc745e85":{"kind":"REWORK","_number":1,"created":"2021-12-14 23:49:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/821762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/821762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/821762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/821762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/821762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/821762/1"}}},"commit":{"parents":[{"commit":"81d845f34e4d350c90af05117fcb0152958b4b3d","subject":"Authors/ChangeLog for 2.23.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81d845f34e4d350c90af05117fcb0152958b4b3d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-13 20:58:32.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-12-14 23:49:52.000000000","tz":0},"subject":"Fix cname_lookup test","message":"Fix cname_lookup test\n\nc.badtest.com actually has a CNAME now, and apparently we\u0027re actually\ndoing the look-up during tests.\n\nChange-Id: I306b7d05740a2b8fcef2f5f432ebf5211bc723cc\n(cherry picked from commit 54fc8a7dee4ad7a38944cbd4c2e3b5f2ec393765)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bffa1e205c499768abd767f0ec4d6f0dc745e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bffa1e205c499768abd767f0ec4d6f0dc745e85"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
