)]}'
{"id":"openstack%2Fswift~821790","triplet_id":"openstack%2Fswift~master~I37e74db28c42d2078a27c543a05b4517ff47f972","project":"openstack/swift","branch":"master","topic":"bucket-ratelimiter","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2021-12-15 22:27:12.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-12-15 16:02:45.000000000","reason":"Alistair Coles replied on the change"}},"hashtags":[],"change_id":"I37e74db28c42d2078a27c543a05b4517ff47f972","subject":"Updater: pop oldest and bucket heapable follow up","status":"ABANDONED","created":"2021-12-15 01:57:00.000000000","updated":"2021-12-15 22:27:12.000000000","total_comment_count":10,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"43eb94f58dac3217f3c00f6f07d8d3ba29b5c99b","_number":821790,"virtual_id_number":821790,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-15 03:05:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-15 03:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-15 10:37:56.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"d7dacc76e406fab6dce1dc6a7a8ccbf8ed4bf939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-15 01:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca85e35f6050f31df57e781ccc4597306f27c07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-15 02:01:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2bad14b27ae539d200bd6852fb42c8ae1f6cd126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 03:05:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96b813083d843ec9fd76713e663bfc5 : SUCCESS in 11m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/ed25c51f67df4b19a2447d414d311129 : SUCCESS in 58m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ede8f7da0aa0405a801ec2be7e25890b : FAILURE in 22m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0120978663e4b67be09587a102719d6 : SUCCESS in 49m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4c8b1aba3acc4000926a54ffe7c58a25 : SUCCESS in 10m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/76e234a9a0c7489b981b3a51ae420a26 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0acf6e532694e7cb702a4d02a90d0f3 : SUCCESS in 17m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed76bbcb225342d19ac77e56faea60f4 : SUCCESS in 20m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7085ca8123d044cbb8777def98be3098 : SUCCESS in 24m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/538455a736ca4534af584756edf2d9da : SUCCESS in 15m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e50c26f71214de99a0be0e137beb993 : SUCCESS in 16m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6943984db514a9db803ce46f1f6ba09 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9dad1b0d27ee41dc90efc55fda9df3d6 : SUCCESS in 22m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d4043eb7b944c0faff5a51f6dee60c7 : SUCCESS in 25m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/78c85ab7c166435f8864ccbf34341d93 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6b471b753d6a42bbb3eaaa91c4ce9722 : SUCCESS in 24m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a609c7d1aeef430eb64ab09c057f3b8e : SUCCESS in 21m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf434d3d069e4c1bb9a86ae3267fa979 : SUCCESS in 8m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0d57afc5505465cb054b908f530563c : SUCCESS in 21m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/902eefb839e64a0d92f3473a619a4026 : FAILURE in 50m 48s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5d20667554e54f058067fabfb6bf4ba2 : SUCCESS in 59m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ece39fe315e446a9914d1407d46b5d9 : SUCCESS in 40m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11fb09bd523b47beb69ad2dcd41138cd : SUCCESS in 47m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fdaa132d9ed482e965edcc6ea81d268 : SUCCESS in 29m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a94136966834cd1a7484f09364bb419 : SUCCESS in 7m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e240e3f0b4046e39b6f2a9e4333ff9f : SUCCESS in 24m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53f6680f899140cf8623d5c604b99443 : FAILURE in 21m 50s","accounts_in_message":[],"_revision_number":2},{"id":"dff23311c14cf14204034254c5ac108f5c102439","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 04:18:51.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1fc7f90402645588107120a3372bc90 : SUCCESS in 50m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24af399bf7c14df3a00be1aa7d4b6780 : SUCCESS in 1h 02m 37s\n- swift-probetests-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/2a151947a0f24781af2dcc92288a31e5 : FAILURE in 2h 05m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/47dda9ce141149fca7b87ea60abd65b3 : SUCCESS in 1h 02m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9078495b293473ab19b29ea07b52662 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1b802088e25a8259c150442ed075cd827086e55f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-15 10:37:56.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0e1272d2fa170dc7e656c22e83bbede9b9290460","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-15 11:02:15.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5a4ac6026dce8fb40f3fcbc02b75df12f34296f2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-15 16:02:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3002cfa3894ea41d02a8fb882d10a66f0bf51f1e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-15 16:24:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43eb94f58dac3217f3c00f6f07d8d3ba29b5c99b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-12-15 22:27:12.000000000","message":"Abandoned\n\nIdeas from this now live in the parent. Can abandon.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"38e488349266464237abed3bfa92bad8d7159b2c","revisions":{"4cea8316113a05ee992b02353d5218a041daa119":{"kind":"REWORK","_number":1,"created":"2021-12-15 01:57:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/90/821790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/821790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/821790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/821790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/821790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/821790/1"}}},"commit":{"parents":[{"commit":"2c72e9a7190aba9d80e80fd72001dbbf5e463018","subject":"WIP updater: per bucket deferral queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c72e9a7190aba9d80e80fd72001dbbf5e463018"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-15 01:49:19.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-15 01:49:19.000000000","tz":660},"subject":"Updater: pop oldest and bucket heapable follow up","message":"Updater: pop oldest and bucket heapable follow up\n\nAdd a __lt__ to the updater Bucket class that compares the\nBucket.wait_until() so it can:\n\n  1. Be used in a heap/priority queue and;\n  2. We now can add buckets directly to the priority queue, no tuples.\n\nAlso change so tracking maxlen only happens in the Bucket iter _not_ the\nBucket class. Also track deferred bucket order so we can also pop the\noldest.\n\nChange-Id: I37e74db28c42d2078a27c543a05b4517ff47f972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cea8316113a05ee992b02353d5218a041daa119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cea8316113a05ee992b02353d5218a041daa119"}]},"branch":"refs/heads/master"},"38e488349266464237abed3bfa92bad8d7159b2c":{"kind":"REWORK","_number":2,"created":"2021-12-15 02:01:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/90/821790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/821790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/821790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/821790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/821790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/821790/2"}}},"commit":{"parents":[{"commit":"2c72e9a7190aba9d80e80fd72001dbbf5e463018","subject":"WIP updater: per bucket deferral queues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c72e9a7190aba9d80e80fd72001dbbf5e463018"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-15 01:49:19.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-12-15 02:00:50.000000000","tz":660},"subject":"Updater: pop oldest and bucket heapable follow up","message":"Updater: pop oldest and bucket heapable follow up\n\nAdd a __lt__ to the updater Bucket class that compares the\nBucket.wait_until() so it can:\n\n  1. Be used in a heap/priority queue and;\n  2. We now can add buckets directly to the priority queue, no tuples.\n\nAlso change so tracking maxlen only happens in the Bucket iter _not_ the\nBucket class. Also track deferred bucket order so we can also pop the\noldest.\n\nChange-Id: I37e74db28c42d2078a27c543a05b4517ff47f972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38e488349266464237abed3bfa92bad8d7159b2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38e488349266464237abed3bfa92bad8d7159b2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
