)]}'
{"id":"openstack%2Fswift~822304","triplet_id":"openstack%2Fswift~master~I9ea43bb079bef5aba0aeee899c224da13d34f918","project":"openstack/swift","branch":"master","topic":"p-test-timeout","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2021-12-20 19:45:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ea43bb079bef5aba0aeee899c224da13d34f918","subject":"Attempt fix test timeouts","status":"MERGED","created":"2021-12-20 10:20:16.000000000","updated":"2021-12-20 19:47:41.000000000","submitted":"2021-12-20 19:45:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822304-p-test-timeout","meta_rev_id":"ee20556000d52add431040a88baae89f7a8cb669","_number":822304,"virtual_id_number":822304,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 19:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-12-20 17:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-12-20 17:31:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-20 11:18:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 17:31:36.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"aec300192d12dde748c22add2256ec3a190c1dd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-20 10:20:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"729f18db7f7eea2d7f8b6cf57085daa753c28a75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 11:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f33805c6570452daf318b3975fa7036 : SUCCESS in 8m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b24098caa8542f78580f4f8fb2fb1e2 : SUCCESS in 48m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/49b48fb503c74ff982c6f837fca3ed25 : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c088e2591bc3487aa8e768dce56384b3 : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee455518e5bd466db639c2abf3b8f8f0 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6bef5ddcd8404093ad799a6b128ddecc : SUCCESS in 15m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af65c716d2ca46ca9fb72d5229e431da : SUCCESS in 18m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ad7b229ea34d4e77bdcfea6ebe00c91e : SUCCESS in 28m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d09b3fa8e58f4e37b58f498b47d4cbed : SUCCESS in 18m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/02949bbd23814268a9a831fe2590356f : SUCCESS in 31m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/516ac268b4f14d0895d57785033bb440 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3453a4746d4540c8ab54a2e3de1c3a4f : SUCCESS in 21m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81cce649821d4a6a86a27fce8f14c369 : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3f50949f00cf4ba6bfef0152dd364f7c : SUCCESS in 18m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2d7b1fa2a1bf42c9b6d278a5965bfdce : SUCCESS in 18m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/58811a6c21974c8d9a7857a686ed3924 : SUCCESS in 46m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5dd2f506868c4adb8e1125e8f59763ab : SUCCESS in 51m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e20513f937814248b4baef0b63cdd080 : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/190c18fc0639490e92e4a15b4522d487 : SUCCESS in 7m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7597da0317a84332809aa25b87772ba9 : SUCCESS in 26m 14s","accounts_in_message":[],"_revision_number":1},{"id":"632c815d0607cfb09a24b339616f658381d16136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-20 12:27:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"71bee5a11bd7bf71cfdc719c3352ba0121987af2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 13:18:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09d7787f14eb4f0bb7155529574ff8f2 : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2039323747d4e1e899116de52e255be : SUCCESS in 38m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b75d1d52def4f97b6ec752130b77775 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8326c44e64d845ccbba0a5e060645e19 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d6049afb87c4477ab9c7425a0d690ec1 : SUCCESS in 18m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/765dc7044ea240c3b55665da1636cbf1 : SUCCESS in 13m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5c32c6faf451439ea4fba808a79fc8fc : SUCCESS in 15m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/050649133aa9461d966be8435f628d06 : SUCCESS in 12m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d00cae6191484f6292d08dfc82690926 : SUCCESS in 16m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/963842ff4a0645aa9e86818e5ef1fcc8 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd23dd11d9124181b70594e1283bcc3e : SUCCESS in 23m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8c01f266b0741e882f59f9765ef8842 : SUCCESS in 16m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e20db4af475a4afeb1823a0c9136b0ff : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f9962e41e07645bc8574c88072883a26 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f0f6b3904d742c58b48ebb906e7c2f8 : SUCCESS in 13m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/af0b95a4dbd34b368b2afbe6cd2ed3ae : SUCCESS in 42m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fa64ff5a257d413f9a29bab2d3263105 : SUCCESS in 44m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7849c0a1d9b49889a0badc9590e3fbb : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3525c5f9254cf1bb9c800e63d1c504 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84676e9319b647a29c2519c37c2f7b54 : SUCCESS in 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"b84220008d644acbe97628d53ce20597bd6d9e58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 14:54:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb3f0cb604f747ed9f63b3b1bd0fc974 : SUCCESS in 40m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bd60b90a6ae44d8992bd89f33173e3e : SUCCESS in 40m 32s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/99c759453dfa48f09ebbc2c192c5277c : SUCCESS in 2h 23m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/292b242b21f64662811f14713a7109df : SUCCESS in 42m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ff92bd2073c42169b3eabc0ff8eb3b4 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f9346216f9c8493cf16e0c4d7ee77a2ed2c1ac00","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-20 17:31:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f70e2cb0a2edc902c052167ec9c0bf5d0c6ff0a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 17:32:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c346cd54b7571c95f39b9ea88af8bea07ccd30e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:45:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2590a77905aa4036804d9a206cf78b81 : SUCCESS in 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f78e1dd8c0f4cfe92690cb330c6aa78 : SUCCESS in 57m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/40da158097c341ba8b7ced5dcdd07f83 : SUCCESS in 9m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/11486777b459496fb6ef772b4297df15 : SUCCESS in 7m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9d8e6796c1d24dec97f19a407f1af8db : SUCCESS in 19m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eb89ada740f843748a5ecbc20229a064 : SUCCESS in 23m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/467c3072a7e7472eb9a7a92c39878803 : SUCCESS in 14m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6b685015455245be9a792f2590a8cb4e : SUCCESS in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/abd5fe436dbd414983677a55d4bcf822 : SUCCESS in 13m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/686b06de773e4796a0dbc20d1dce3ec6 : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/15e19e36dcd74a8d9195f8188be23e5c : SUCCESS in 15m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/44d24191d6594ef696255736f7530f14 : SUCCESS in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/651fcee9b19c43338b958232bcd7d394 : SUCCESS in 13m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2dfaec9ba28c455fb12331bd71091997 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99dc5e817dad41ddb4915d7bf68f0beb : SUCCESS in 22m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29166f665c7545f5b3ae46a72e4aa795 : SUCCESS in 7m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c8364b4e4614f7f872caa58d3eca595 : SUCCESS in 42m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70fc93c7b1e5440ebdf0a326eec8fb35 : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4641e7059ead45299515e1ed74e4b60f : SUCCESS in 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057a05c8839e4fb2b5b7b0d018791798 : SUCCESS in 8m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f5f207021c74f038196c996d83543dc : SUCCESS in 20m 56s","accounts_in_message":[],"_revision_number":2},{"id":"18a0b722c857534ce4a9678b05e12a05a9be7bbf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:45:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ee20556000d52add431040a88baae89f7a8cb669","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 19:47:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776699b9f4a14b799cd1fcfd97db3a7f : SUCCESS in 1m 31s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/df6f2fa6dedc4d8987c4bbb34d5c978e : SUCCESS in 54s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0bd7db36f05b4c7db6880f62a0b4ff83 : SUCCESS in 1m 19s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b1f03149f03d25d1808bf11c6fb4e8f2e4dca61f","revisions":{"c9987cf069b026b7d9780de6ae01c20c1c60c831":{"kind":"REWORK","_number":1,"created":"2021-12-20 10:20:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/04/822304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/822304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/822304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/822304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/822304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/822304/1"}}},"commit":{"parents":[{"commit":"c6f37573cfd3e8da8b0c765f761b0d46f410179c","subject":"Modify log_name in internal clients\u0027 pipeline configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6f37573cfd3e8da8b0c765f761b0d46f410179c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-20 10:16:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-20 10:16:05.000000000","tz":0},"subject":"WIP maybe fix test timeouts","message":"WIP maybe fix test timeouts\n\nZuul unit test jobs have been timimg out, often while executing a test\nthat attempts getaddrinfo. Mock the getaddrinfo to see if that helps.\n\nChange-Id: I9ea43bb079bef5aba0aeee899c224da13d34f918\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9987cf069b026b7d9780de6ae01c20c1c60c831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9987cf069b026b7d9780de6ae01c20c1c60c831"}]},"branch":"refs/heads/master"},"b1f03149f03d25d1808bf11c6fb4e8f2e4dca61f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2021-12-20 12:27:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/04/822304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/822304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/822304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/822304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/822304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/822304/2"}}},"commit":{"parents":[{"commit":"ebf75f6e2a7e82a38757e7e214c3f321a5a4d10e","subject":"Merge \"Move CI from CentOS 8 to CentOS 8 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebf75f6e2a7e82a38757e7e214c3f321a5a4d10e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-20 10:16:05.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-12-20 12:26:35.000000000","tz":0},"subject":"Attempt fix test timeouts","message":"Attempt fix test timeouts\n\nZuul unit test jobs have sometimes been timing out, often while\nexecuting a test that attempts getaddrinfo. Mock the getaddrinfo call\nto see if that helps.\n\nChange-Id: I9ea43bb079bef5aba0aeee899c224da13d34f918\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1f03149f03d25d1808bf11c6fb4e8f2e4dca61f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1f03149f03d25d1808bf11c6fb4e8f2e4dca61f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
