)]}'
{"id":"openstack%2Fswift~822377","triplet_id":"openstack%2Fswift~master~I5cfabb7255567d80160f9096ddaee85ac9ed5cfc","project":"openstack/swift","branch":"master","topic":"object-expire-mpu","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-01-11 16:31:42.000000000","reason":"Alistair Coles replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5cfabb7255567d80160f9096ddaee85ac9ed5cfc","subject":"WIP: use async pending to expire MPU","status":"NEW","created":"2021-12-21 02:30:16.000000000","updated":"2022-01-11 16:31:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":12,"unresolved_comment_count":10,"has_review_started":true,"meta_rev_id":"04798ee8bacfb2183a0ee8209b59e899aef201e3","_number":822377,"virtual_id_number":822377,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-07 23:12:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-21 04:28:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-21 23:20:33.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-12-22 14:47:14.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"492e3253b9247a0f4f92001b44b651bede2db053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-12-21 02:30:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de67179a419099814deb6d16aa62d5842f03d4b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 04:28:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db8b9c9dc79e4327a5b3a8ba71081ee7 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/fd144bc7c4ec476e92fe71cf67d5b2c9 : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/273c6a2e7abd425b8ab02eadd7533a41 : SUCCESS in 49m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4d2b10085ec4ee283c7977137424240 : SUCCESS in 50m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3bbae1f8222344d8955c0d59c1565e4d : SUCCESS in 11m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e5210b5271514125bac7d2b7fed1b9a0 : SUCCESS in 8m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b25deddd4d5b4750953c694327fb7d29 : FAILURE in 21m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d394a91ecd54c51a69040d9c0722371 : TIMED_OUT in 1h 32m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c0e046afd3144ba48f18b7a50255d57b : FAILURE in 20m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/04d2eed29fbe44ae996f6a3a46e6afcf : FAILURE in 25m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d8fafcd5ce034e4cae726088588b2b15 : FAILURE in 18m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/310902c034984b3da0e19e3fb3daf2a5 : SUCCESS in 22m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e781817be5784c0c93eaa6d3cada63d2 : SUCCESS in 20m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/16c16ded76a14d6c88dc222f77fe9d05 : SUCCESS in 21m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/69fbf998cdd44ceb8717d3795bf8d98f : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/34b404549c294022acb5cd6482f62290 : SUCCESS in 28m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/564b60f272794ff89cfec5ca4e35de2f : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/574991bf3ff7439a98c6bb8eb00d4046 : SUCCESS in 8m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff71824ff7a54e70bbe960421bdc410e : SUCCESS in 25m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2883a9509d384b0f9474674a14a3adc6 : SUCCESS in 57m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac5426548f204b83a3605373d3cda156 : FAILURE in 1h 07m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6fc327e262cf49889be4568d7877f8d4 : SUCCESS in 44m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a91d9eef318d4d2e9c13843e5a10c68b : SUCCESS in 42m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0344c8878af14ff68d0d9ad5795626e9 : FAILURE in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e62e421f88674711ba5649fcea85f930 : SUCCESS in 8m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55c019fe1cf9404a96a13b3400149383 : SUCCESS in 24m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6486b13c078d44d7a65483c238816ca9 : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b13ba140d426d3288fd8f152ed6de67c8a2dbdc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 05:45:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0d4cf01c3514d32bcafb1062dffcd6e : FAILURE in 1h 09m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/262cd32f3e2f49a4963a54bf4534a7e8 : TIMED_OUT in 1h 31m 51s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b783e3d0840f45559dbd8cc774e145d6 : FAILURE in 2h 54m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/841998ad6c4d4145bdecec0d249e19ab : SUCCESS in 1h 11m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab852db706de4bfc8fa8916ba2cf321d : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c616a1911c770ffa4bbda7b9175573deaf8f3315","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-21 23:20:33.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5604cc871f04485b9b4938ea914915251cff0f51","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-12-22 14:47:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9b7408c873b529d032223cdb1281cb0094a0be0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-07 22:06:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d27ab2dda9663d033b4f13625437659efe83b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 23:12:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51e2c079720f4f70a69224daf6054f45 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/3416c18ecfda4922b8b92bb17004a5db : SUCCESS in 55m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8442f4ac6cf34b758eaeeecf0ae250c5 : SUCCESS in 44m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa403cca0a324296992633e239e2183a : SUCCESS in 46m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd19476aa0f34ae5a779519c6b0f8568 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12d5222a88f245abae7e4f6c24e51b48 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ddcaf94cc1d4f60abfae1de328ee90f : FAILURE in 23m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/069bea926b0b480297cb73d4d2981d95 : FAILURE in 25m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b0205c95c04430cbbd001bce0c955c6 : FAILURE in 23m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c766c9c3e36646ad91b34006617370d5 : SUCCESS in 33m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8748f434c2d144328a56591108f203d5 : SUCCESS in 25m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1043af9d42ee44a282a1d91a1f14c73e : SUCCESS in 22m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/87d24559d74f4130beb6df60729443d4 : SUCCESS in 27m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1613561cb313406aa1de7e1c5a78ffd6 : SUCCESS in 22m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c539558459b45df984976ece98bff69 : SUCCESS in 30m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1600db2229aa4430a76a112a399098df : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e748aceb29ce4e2fb8dd01301d4f78a4 : SUCCESS in 25m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2524e6a5c9a4184bd633623a7c1918c : SUCCESS in 52m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/96590f03d7b644daa23449e1f96e5b07 : SUCCESS in 56m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ba496d0b45e4931893fd55216e9aed3 : SUCCESS in 41m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c86d7cd89ce44545b1f54ad9e06387f7 : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c48d55e1e214a7280d6187bb7b4c624 : FAILURE in 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50aa15fc0e2467381a798887327223a : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28ef2cbd814745a88e51767d8a709e4e : SUCCESS in 29m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad44b2785a0249a4967f7ae9f9d095c2 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":2},{"id":"990a3373b4cc1daee118f65554374e7867eaeafc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 01:40:49.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb0fdc5a52cb4331909f6280e364d2cd : FAILURE in 35m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adca3ea346e04d1fb360cab0853e6125 : FAILURE in 38m 40s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/08144ace3d6243d9bf15bb95cb272ba0 : FAILURE in 2h 22m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ed61cf2da374064a2b4846448713bc1 : SUCCESS in 50m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c27f40e1bdd4f1cbf26f48da1236405 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b1410b6caa3827a31e605f437b3e3bbf6fc3c478","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-01-10 23:23:52.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"04798ee8bacfb2183a0ee8209b59e899aef201e3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-01-11 16:31:42.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"48b5a13344f6b203bc879608c1d0139c5e643afd","revisions":{"63f39ce70658fa95f82f553f9975cc36ba0cc024":{"kind":"REWORK","_number":1,"created":"2021-12-21 02:30:16.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/77/822377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/822377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/822377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/822377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/822377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/822377/1"}}},"commit":{"parents":[{"commit":"768dc510d29295651bba1f1542cb2e36ada0d301","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/768dc510d29295651bba1f1542cb2e36ada0d301"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-21 02:27:12.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-21 02:28:20.000000000","tz":-360},"subject":"WIP: use async pending to expire MPU","message":"WIP: use async pending to expire MPU\n\nThis is still just a best effort cleanup, s3api MPUs are known to\ngenerate orphaned segments [1], but if if this turns out to be\nreasonable it might also be useful for MPU overwrite.\n\n1. lp bug #1813202\n\nChange-Id: I5cfabb7255567d80160f9096ddaee85ac9ed5cfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63f39ce70658fa95f82f553f9975cc36ba0cc024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63f39ce70658fa95f82f553f9975cc36ba0cc024"}]},"branch":"refs/heads/master"},"48b5a13344f6b203bc879608c1d0139c5e643afd":{"kind":"REWORK","_number":2,"created":"2022-01-07 22:06:25.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/77/822377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/822377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/822377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/822377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/822377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/822377/2"}}},"commit":{"parents":[{"commit":"3972b8b4baf20a36d140c90d6e38316e5d839104","subject":"DNM: add a probetest for MPU expiration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3972b8b4baf20a36d140c90d6e38316e5d839104"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-12-21 02:27:12.000000000","tz":-360},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-01-07 22:05:27.000000000","tz":-360},"subject":"WIP: use async pending to expire MPU","message":"WIP: use async pending to expire MPU\n\nThis is still just a best effort cleanup, s3api MPUs are known to\ngenerate orphaned segments [1], but if if this turns out to be\nreasonable it might also be useful for MPU overwrite.\n\n1. lp bug #1813202\n\nChange-Id: I5cfabb7255567d80160f9096ddaee85ac9ed5cfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48b5a13344f6b203bc879608c1d0139c5e643afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48b5a13344f6b203bc879608c1d0139c5e643afd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
